Skip to content

Refactor Link component to use CSS modules - #4828

Merged
jonrohan merged 12 commits into
mainfrom
css_modules_convert_link
Aug 14, 2024
Merged

Refactor Link component to use CSS modules#4828
jonrohan merged 12 commits into
mainfrom
css_modules_convert_link

Conversation

@jonrohan

@jonrohanjonrohan commented Aug 8, 2024

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3719

Changed

  • Update Link component to use CSS Modules
  • Add feature flag to Link component to toggle between styled-components and CSS Modules
  • Use feature flag for VRT in Link

Removed

Rollout strategy

  • Minor release

Testing & Reviewing

@changeset-bot

changeset-botBot commented Aug 8, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6f10e4e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 8, 2024
@github-actions

github-actionsBot commented Aug 8, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.13 KB (-0.08% 🔽)
packages/react/dist/browser.umd.js96.48 KB (+0.03% 🔺)

joshblackand others added 2 commits August 12, 2024 20:13
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4828 August 13, 2024 23:35 Inactive
@jonrohan
jonrohan marked this pull request as ready for review August 13, 2024 23:35
@jonrohan
jonrohan requested a review from a team as a code ownerAugust 13, 2024 23:35

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice

@jonrohan
jonrohan added this pull request to the merge queueAug 14, 2024
Merged via the queue into main with commit 1a674f7Aug 14, 2024
@jonrohan
jonrohan deleted the css_modules_convert_link branch August 14, 2024 22:18
@primerprimerBot mentioned this pull request Aug 14, 2024
jonrohan added a commit that referenced this pull request Aug 15, 2024
* Refactor Link to use CSS modules
* Remove needless disable
* Add newline
* Create quick-adults-buy.md
* Run tests inside fhte feature flag also
* Fix naming
* Rails interpolation got me
* test(e2e): update e2e tests for link (#4825)
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Remove example-nextjs
---------
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@joshblack@langermank