Skip to content

refactor(Text): Text component to use CSS modules behind primer_react_css_modules_team feature flag - #4874

Merged
jonrohan merged 4 commits into
mainfrom
primer_css_modules_text
Aug 30, 2024
Merged

refactor(Text): Text component to use CSS modules behind primer_react_css_modules_team feature flag#4874
jonrohan merged 4 commits into
mainfrom
primer_css_modules_text

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3785

Changelog

Changed

Refactoring Text to use CSS modules behind the primer_react_css_modules_team feature flag.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

This will be behind the primer_react_css_modules_team feature flag

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerAugust 22, 2024 20:22
@changeset-bot

changeset-botBot commented Aug 22, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5628650

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 22, 2024
@github-actions

github-actionsBot commented Aug 22, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.39 KB (+0.07% 🔺)
packages/react/dist/browser.umd.js96.73 KB (+0.09% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4874 August 22, 2024 20:27 Inactive
@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 22, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Aug 22, 2024
@jonrohanjonrohan changed the title refactor(Text): To CSS modules behind primer_react_css_modules_team feature flagrefactor(Text): Text component to use CSS modules behind primer_react_css_modules_team feature flagAug 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4874 August 22, 2024 21:46 Inactive
id: story.id,
globals: {
featureFlags: {
primer_react_css_modules_team: true,

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Quick question: Does this mean we only run these tests with the feature flag on? Or do we run it twice, once with and once without

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Or do we run it twice, once with and once without

Running once with it off and once with it on and comparing them to the same screenshot.

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Ooh that's smart!

@jonrohan
jonrohan added this pull request to the merge queueAug 30, 2024
Merged via the queue into main with commit 4c69b38Aug 30, 2024
@jonrohan
jonrohan deleted the primer_css_modules_text branch August 30, 2024 02:00
@primerprimerBot mentioned this pull request Aug 29, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@siddharthkp