Skip to content

chore(Text): Remove CSS modules feature flag from Text component - #5061

Closed
jonrohan wants to merge 19 commits into
mainfrom
css_modules_remove_feature_flag_text
Closed

chore(Text): Remove CSS modules feature flag from Text component#5061
jonrohan wants to merge 19 commits into
mainfrom
css_modules_remove_feature_flag_text

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 4, 2024

Copy link
Copy Markdown
Member

This removes the CSS modules feature flag from the Text component that was first introduced 2 months ago #4874

Changelog

Changed

Text is styled by CSS modules.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Oct 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0ed7a78

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5061 October 4, 2024 00:02 Inactive
@github-actions

github-actionsBot commented Oct 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.36 KB (-0.05% 🔽)
packages/react/dist/browser.umd.js104.73 KB (-0.14% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5061 October 7, 2024 23:05 Inactive
@jonrohan
jonrohan marked this pull request as ready for review October 8, 2024 18:23
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 8, 2024 18:23
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 15, 2024

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🔥

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5061 November 13, 2024 19:29 Inactive
@jonrohan
jonrohan marked this pull request as draft November 14, 2024 20:10
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/359515

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@jonrohan
jonrohan deleted the css_modules_remove_feature_flag_text branch May 9, 2025 18:32
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank