Skip to content

Bug fix: Button loading spinner color - #4941

Merged
langermank merged 4 commits into
mainfrom
fix-button-spinner
Sep 9, 2024
Merged

Bug fix: Button loading spinner color#4941
langermank merged 4 commits into
mainfrom
fix-button-spinner

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Noticed this loading spinner color is wrong as I'm rewriting this component with CSS Modules. Fixing it here to fix my new branch 😄

@changeset-bot

changeset-botBot commented Sep 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a980a26

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 9, 2024
@github-actions

github-actionsBot commented Sep 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.56 KB (-0.08% 🔽)
packages/react/dist/browser.umd.js96.88 KB (-0.03% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4941 September 9, 2024 20:42 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 9, 2024
@langermank
langermank marked this pull request as ready for review September 9, 2024 20:47
@langermank
langermank requested review from a team as code ownersSeptember 9, 2024 20:47
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4941 September 9, 2024 20:51 Inactive

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🎉

@langermank
langermank added this pull request to the merge queueSep 9, 2024
Merged via the queue into main with commit 80fe025Sep 9, 2024
@langermank
langermank deleted the fix-button-spinner branch September 9, 2024 22:01
@primerprimerBot mentioned this pull request Sep 9, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan