Skip to content

Test PR (Tooltip) - #4961

Closed
TylerJDev wants to merge 7 commits into
mainfrom
tooltip-test
Closed

Test PR (Tooltip)#4961
TylerJDev wants to merge 7 commits into
mainfrom
tooltip-test

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Sep 13, 2024

Copy link
Copy Markdown
Member

Moved #4967

Changelog

Changed

  • Uses [disabled] instead of :disabled

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0ad1542

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Sep 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.42 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js97.69 KB (-0.04% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4961 September 13, 2024 21:21 Inactive
@TylerJDevTylerJDev mentioned this pull request Sep 13, 2024
13 tasks
@primer-integration

Copy link
Copy Markdown

😢 Hi from github/github. The integration workflow has failed: https://github.com/github/github/actions/runs/10877125471

@TylerJDevTylerJDev changed the title Test PR (Tooltip)TooltipV2: Use [disabled] selector to check for disabled elementsSep 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4961 September 16, 2024 12:55 Inactive
@TylerJDevTylerJDev changed the title TooltipV2: Use [disabled] selector to check for disabled elementsTest PR (Tooltip)Sep 16, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@TylerJDev@broccolinisoup@langermank@lukasoppermann