Skip to content

TooltipV2: Use [disabled] selector to check for disabled elements - #4967

Merged
TylerJDev merged 1 commit into
mainfrom
tooltip-add-selector
Sep 17, 2024
Merged

TooltipV2: Use [disabled] selector to check for disabled elements#4967
TylerJDev merged 1 commit into
mainfrom
tooltip-add-selector

Conversation

@TylerJDev

Copy link
Copy Markdown
Member

A bug in Jest was causing a false positive when using :disabled, this is now fixed by using [disabled] instead.

Context for the false positive: dperini/nwsapi#125

Changelog

Changed

  • Uses [disabled] instead of :disabled

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@TylerJDev
TylerJDev requested a review from a team as a code ownerSeptember 16, 2024 12:57
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 028e686

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.42 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js97.69 KB (-0.04% 🔽)

@TylerJDevTylerJDev changed the title TooltipV2: Use [disabled] selector to check for disabled elements #4961TooltipV2: Use [disabled] selector to check for disabled elementsSep 16, 2024
@TylerJDevTylerJDev mentioned this pull request Sep 16, 2024
13 tasks

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚀

@TylerJDev
TylerJDev added this pull request to the merge queueSep 17, 2024
Merged via the queue into main with commit 98d3d13Sep 17, 2024
@TylerJDev
TylerJDev deleted the tooltip-add-selector branch September 17, 2024 20:48
@primerprimerBot mentioned this pull request Sep 17, 2024
TylerJDev added a commit that referenced this pull request Sep 23, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@broccolinisoup