Skip to content

Popover: Remove the caret - #4978

Closed
TylerJDev wants to merge 6 commits into
mainfrom
remove-the-caret
Closed

Popover: Remove the caret#4978
TylerJDev wants to merge 6 commits into
mainfrom
remove-the-caret

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Sep 17, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3432

Removes the caret in Popover.

Changelog

Removed

  • Removes caret prop and associated styles from Popover component

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 17, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8f80d76

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Sep 17, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.12 KB (-0.36% 🔽)
packages/react/dist/browser.umd.js97.46 KB (-0.23% 🔽)

@TylerJDevTylerJDev changed the title Remove the caretPopover: Remove the caretSep 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4978 September 18, 2024 13:31 Inactive
@TylerJDevTylerJDev mentioned this pull request Sep 20, 2024
13 tasks
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4978 September 20, 2024 17:39 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 20, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 20, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4978 September 20, 2024 17:55 Inactive
@TylerJDev

Copy link
Copy Markdown
MemberAuthor

Hey @joshblack! With caret being deprecated, I wanted to create a PR that removes it entirely (in v38). I'm curious if the best course of action is to wait until v37 is fully shipped, then to open this PR up for reviews/merging?

@joshblack

Copy link
Copy Markdown
Member

Yeah that works @TylerJDev! We did a similar thing for MarkdownEditor this time and I bet it would work for this too

@TylerJDev
TylerJDev marked this pull request as ready for review October 21, 2024 22:44
@TylerJDev
TylerJDev requested review from a team as code ownersOctober 21, 2024 22:44
@joshblack

Copy link
Copy Markdown
Member

@TylerJDev for timing on this, would it be helpful to open it when we enter into "pre mode" for v38? Then we could merge it in as-is and go through that route. I think I misspoke earlier when I said open after v37 is merged since this would need to be in the RC for v38 versus being able to land in v37

@TylerJDev

Copy link
Copy Markdown
MemberAuthor

@TylerJDev for timing on this, would it be helpful to open it when we enter into "pre mode" for v38? Then we could merge it in as-is and go through that route. I think I misspoke earlier when I said open after v37 is merged since this would need to be in the RC for v38 versus being able to land in v37

Makes sense! I'll put this back into draft until then 😄

@TylerJDev
TylerJDev marked this pull request as draft October 22, 2024 20:21
@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions
github-actionsBot deleted the remove-the-caret branch April 29, 2025 20:05
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@TylerJDev@joshblack@lesliecdubs