Skip to content

Popover: Deprecate the caret - #4997

Merged
TylerJDev merged 3 commits into
mainfrom
deprecate-the-caret
Sep 20, 2024
Merged

Popover: Deprecate the caret#4997
TylerJDev merged 3 commits into
mainfrom
deprecate-the-caret

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Sep 20, 2024

Copy link
Copy Markdown
Member

Prerequisite to #4978

Changelog

Changed

  • Add comment on deprecation of Caret in v38

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 20, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5458f7b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Sep 20, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.37 KB (0%)
packages/react/dist/browser.umd.js97.79 KB (0%)

@TylerJDev
TylerJDev added this pull request to the merge queueSep 20, 2024
Merged via the queue into main with commit dc2d1f3Sep 20, 2024
@TylerJDev
TylerJDev deleted the deprecate-the-caret branch September 20, 2024 15:36
@primerprimerBot mentioned this pull request Sep 20, 2024
TylerJDev added a commit that referenced this pull request Sep 23, 2024
* deprecate the caret
* Fix comment
* Add changeset
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack