Skip to content

chore(Label): Remove CSS modules feature flag from Label component - #5060

Merged
jonrohan merged 12 commits into
mainfrom
remove_feature_flag_label
Oct 24, 2024
Merged

chore(Label): Remove CSS modules feature flag from Label component#5060
jonrohan merged 12 commits into
mainfrom
remove_feature_flag_label

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 3, 2024

Copy link
Copy Markdown
Member

This removes the CSS modules feature flag from the Label component that was first introduced 4 weeks ago #4893

Changelog

Changed

Label is styled by CSS modules.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Oct 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2bef3a1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5060 October 4, 2024 00:01 Inactive
@github-actions

github-actionsBot commented Oct 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.96 KB (-0.17% 🔽)
packages/react/dist/browser.umd.js97.21 KB (-0.31% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5060 October 7, 2024 23:06 Inactive
@jonrohan
jonrohan marked this pull request as ready for review October 8, 2024 17:45
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 8, 2024 17:45
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 15, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5060 October 15, 2024 23:07 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/348598

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@jonrohan
jonrohan added this pull request to the merge queueOct 24, 2024
Merged via the queue into main with commit 33396eaOct 24, 2024
@jonrohan
jonrohan deleted the remove_feature_flag_label branch October 24, 2024 15:42
@primerprimerBot mentioned this pull request Oct 24, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack