Skip to content

Bug fix(Avatar): Changed rounded border calculation for the "square" Avatar to better align with existing border styles - #5077

Merged
jonrohan merged 2 commits into
mainfrom
avatar_rounded_border_fix
Oct 7, 2024
Merged

Bug fix(Avatar): Changed rounded border calculation for the "square" Avatar to better align with existing border styles#5077
jonrohan merged 2 commits into
mainfrom
avatar_rounded_border_fix

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 7, 2024

Copy link
Copy Markdown
Member

Fixes an issue I noticed in #5070 where the rounded corners of the "square" Avatar don't match the unflagged version.

You can see the problem this change is trying to fix in the Header storybook example with flag off vs with staff flag.

Changelog

Changed

Changed rounded border calculation for the "square" Avatar to better align with existing border styles

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerOctober 7, 2024 17:28
@changeset-bot

changeset-botBot commented Oct 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f43f926

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.93 KB (0%)
packages/react/dist/browser.umd.js97.12 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5077 October 7, 2024 17:31 Inactive
@jonrohan

jonrohan commented Oct 7, 2024

Copy link
Copy Markdown
MemberAuthor

@langermank you can see the difference here if you look at Header storybook with flag off vs with staff flag.

@jonrohan
jonrohan enabled auto-merge October 7, 2024 17:33

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice use of clamp!

@jonrohan
jonrohan added this pull request to the merge queueOct 7, 2024
Merged via the queue into main with commit 6490b27Oct 7, 2024
@jonrohan
jonrohan deleted the avatar_rounded_border_fix branch October 7, 2024 18:57
@primerprimerBot mentioned this pull request Oct 7, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank