Skip to content

Bug Fix for Text component and styled system arguments - #5079

Merged
jonrohan merged 5 commits into
mainfrom
bug_fix_text
Oct 7, 2024
Merged

Bug Fix for Text component and styled system arguments#5079
jonrohan merged 5 commits into
mainfrom
bug_fix_text

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 7, 2024

Copy link
Copy Markdown
Member

This was discovered in #5070. The Text component wasn't doing anything with styled system typography or common props. This PR adds logic to ensure that these props are correctly passed to the Box component when the Text component is used with the CSS modules feature flag enabled.

You can observe the bug by visiting the SegmentedControl storybook example with flag off vs flag on.

Changelog

Changed

Correctly pass styled system typography and common props to the Box component in the Text component when the CSS modules feature flag is enabled.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerOctober 7, 2024 17:47
@changeset-bot

changeset-botBot commented Oct 7, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e247b44

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 7, 2024
@jonrohan
jonrohan enabled auto-merge October 7, 2024 17:47
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5079 October 7, 2024 17:50 Inactive
@github-actions

github-actionsBot commented Oct 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.92 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js97.16 KB (+0.05% 🔺)

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Left a comment for the helper function, was curious what you would say 🤔 Wasn't sure if this was in a fast path or not / if this component can be used a lot on a page.

Comment threadpackages/react/src/Text/Text.tsx
Co-authored-by: Josh Black <joshblack@github.com>
@jonrohan
jonrohan disabled auto-merge October 7, 2024 17:54
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5079 October 7, 2024 17:57 Inactive
@jonrohan
jonrohan added this pull request to the merge queueOct 7, 2024
Merged via the queue into main with commit a2e5671Oct 7, 2024
@jonrohan
jonrohan deleted the bug_fix_text branch October 7, 2024 19:27
@primerprimerBot mentioned this pull request Oct 7, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack