Skip to content

feat(TextInput): Convert TextInput (Part 3) - Update TextInputWrapper and TextInputWithTokens - #5227

Merged
hussam-i-am merged 78 commits into
mainfrom
css-moduled-text-input
Nov 14, 2024
Merged

feat(TextInput): Convert TextInput (Part 3) - Update TextInputWrapper and TextInputWithTokens#5227
hussam-i-am merged 78 commits into
mainfrom
css-moduled-text-input

Conversation

@hussam-i-am

@hussam-i-amhussam-i-am commented Nov 4, 2024

Copy link
Copy Markdown
Contributor

Changelog

New

Changed

  • Updated TextInputWrapper to use CSS modules behind feature flag
  • Partially updated TextInputWithTokens to use CSS modules for TextInputWrapper
  • Updated snapshots

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@hussam-i-amhussam-i-am self-assigned this Nov 4, 2024
@changeset-bot

changeset-botBot commented Nov 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 96c2dc4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5227 November 4, 2024 22:05 Inactive
@github-actions

github-actionsBot commented Nov 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js99.49 KB (+0.17% 🔺)
packages/react/dist/browser.umd.js99.81 KB (+0.15% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/351001

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

Base automatically changed from revert-5256-revert-5242-hussam-i-am/text-input-data-attrs to mainNovember 14, 2024 18:05
@hussam-i-am
hussam-i-am added this pull request to the merge queueNov 14, 2024
Merged via the queue into main with commit 1cfbf63Nov 14, 2024
@primerprimerBot mentioned this pull request Nov 14, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@hussam-i-am@jonrohan