Skip to content

feat(Tooltip): Convert Tooltip to CSS modules behind team flag - #5228

Merged
jonrohan merged 6 commits into
mainfrom
css_modules_tooltip
Nov 7, 2024
Merged

feat(Tooltip): Convert Tooltip to CSS modules behind team flag#5228
jonrohan merged 6 commits into
mainfrom
css_modules_tooltip

Conversation

@jonrohan

@jonrohanjonrohan commented Nov 4, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4296

Changelog

Changed

Convert Tooltip to CSS modules behind team flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Nov 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 72bc2ad

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 4, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5228 November 4, 2024 22:35 Inactive
Co-authored-by: Keith Cirkel <keithamus@users.noreply.github.com>
@github-actions

github-actionsBot commented Nov 7, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.5 KB (+0.1% 🔺)
packages/react/dist/browser.umd.js98.8 KB (+0.14% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/350441

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 7, 2024
@jonrohan
jonrohan marked this pull request as ready for review November 7, 2024 20:52
@jonrohan
jonrohan requested a review from a team as a code ownerNovember 7, 2024 20:52
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5228 November 7, 2024 22:08 Inactive
@jonrohan
jonrohan added this pull request to the merge queueNov 7, 2024
Merged via the queue into main with commit 2da7336Nov 7, 2024
@jonrohan
jonrohan deleted the css_modules_tooltip branch November 7, 2024 23:14
@primerprimerBot mentioned this pull request Nov 7, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank