Skip to content

Bring back "SelectPanel: Prepare for non-anchored variants #5230" - #5273

Closed
siddharthkp wants to merge 1 commit into
mainfrom
revert-5255-revert-5230-selectpanel-use-anchored-position
Closed

Bring back "SelectPanel: Prepare for non-anchored variants #5230"#5273
siddharthkp wants to merge 1 commit into
mainfrom
revert-5255-revert-5230-selectpanel-use-anchored-position

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

@siddharthkpsiddharthkp self-assigned this Nov 13, 2024
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2d6fa8f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.8 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js99.18 KB (+0.2% 🔺)

@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actions
github-actionsBot deleted the revert-5255-revert-5230-selectpanel-use-anchored-position branch February 5, 2025 04:05
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff memberStale

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@siddharthkp