Skip to content

refactor(FormControl): update to CSS Modules behind feature flag - #5370

Merged
joshblack merged 11 commits into
mainfrom
refactor/update-form-control-to-css-modules-3
Dec 11, 2024
Merged

refactor(FormControl): update to CSS Modules behind feature flag#5370
joshblack merged 11 commits into
mainfrom
refactor/update-form-control-to-css-modules-3

Conversation

@joshblack

@joshblackjoshblack commented Dec 5, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3812

Changelog

New

Changed

  • Update FormControl and shared internal components to use form control behind feature flag

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Dec 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6491edb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 5, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 5, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.07 KB (+0.34% 🔺)
packages/react/dist/browser.umd.js103.44 KB (+0.29% 🔺)

Comment threadpackages/react/src/FormControl/FormControl.module.css Outdated
Comment threadpackages/react/src/FormControl/FormControl.module.css Outdated
Comment threadpackages/react/src/internal/components/InputLabel.module.css Outdated
Comment threadpackages/react/src/internal/components/InputLabel.module.css Outdated
Comment threadpackages/react/src/internal/components/InputValidation.module.css Outdated
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5370 December 6, 2024 17:06 Inactive
@joshblack
joshblack marked this pull request as ready for review December 9, 2024 16:08
@joshblack
joshblack requested a review from a team as a code ownerDecember 9, 2024 16:08

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Woo, here we go 🎉

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/354624

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 11, 2024
@joshblack
joshblack added this pull request to the merge queueDec 11, 2024
Merged via the queue into main with commit 204ded0Dec 11, 2024
@joshblack
joshblack deleted the refactor/update-form-control-to-css-modules-3 branch December 11, 2024 18:13
@primerprimerBot mentioned this pull request Dec 11, 2024
francinelucca added a commit that referenced this pull request Dec 13, 2024
github-merge-queueBot pushed a commit that referenced this pull request Dec 13, 2024
* Revert "refactor(FormControl): update to CSS Modules behind feature flag (#5370)"
This reverts commit 204ded0.
* Revert "refactor(Table): update to CSS Modules (#5349)"
This reverts commit 2c79798.
* Create healthy-eggs-pretend.md
@joshblack
joshblack restored the refactor/update-form-control-to-css-modules-3 branch December 13, 2024 16:09
@joshblack
joshblack deleted the refactor/update-form-control-to-css-modules-3 branch December 13, 2024 16:09
@joshblack
joshblack restored the refactor/update-form-control-to-css-modules-3 branch December 13, 2024 17:18
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5438 December 13, 2024 17:20 Inactive
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@langermank