Skip to content

Revert #5370, #5349 due to visual regressions - #5434

Merged
francinelucca merged 3 commits into
mainfrom
revert-5370-refactor/update-form-control-to-css-modules-3
Dec 13, 2024
Merged

Revert #5370, #5349 due to visual regressions#5434
francinelucca merged 3 commits into
mainfrom
revert-5370-refactor/update-form-control-to-css-modules-3

Conversation

@francinelucca

@francineluccafrancinelucca commented Dec 13, 2024

Copy link
Copy Markdown
Member

@changeset-bot

changeset-botBot commented Dec 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bdfb1c0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.21 KB (-0.31% 🔽)
packages/react/dist/browser.umd.js103.55 KB (-0.27% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/355040

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Dec 13, 2024
@francinelucca
francinelucca marked this pull request as ready for review December 13, 2024 14:42
@francinelucca
francinelucca requested a review from a team as a code ownerDecember 13, 2024 14:42
@francineluccafrancinelucca changed the title Revert "refactor(FormControl): update to CSS Modules behind feature flag"Revert #5370, #5349 due to visual regressionsDec 13, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@francineluccafrancinelucca added status: review needed and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@francinelucca
francinelucca added this pull request to the merge queueDec 13, 2024
Merged via the queue into main with commit da7debdDec 13, 2024
@francinelucca
francinelucca deleted the revert-5370-refactor/update-form-control-to-css-modules-3 branch December 13, 2024 15:09
@primerprimerBot mentioned this pull request Dec 13, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff memberstatus: review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@TylerJDev