Skip to content

refactor(FormControl): update FormControl to use CSS Modules behind flag - #5438

Merged
jonrohan merged 17 commits into
mainfrom
refactor/update-form-control-to-css-modules-3
Dec 19, 2024
Merged

refactor(FormControl): update FormControl to use CSS Modules behind flag#5438
jonrohan merged 17 commits into
mainfrom
refactor/update-form-control-to-css-modules-3

Conversation

@joshblack

@joshblackjoshblack commented Dec 13, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3812

Changelog

New

Changed

  • Update FormControl and shared internal components to use form control behind feature flag

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Dec 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bb0a9da

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.69 KB (+0.35% 🔺)
packages/react/dist/browser.umd.js104 KB (+0.32% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5438 December 13, 2024 17:37 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@hussam-i-amhussam-i-am self-assigned this Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5438 December 18, 2024 18:51 Inactive
@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Dec 18, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5438 December 18, 2024 21:14 Inactive
@github-actionsgithub-actionsBot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@hussam-i-am
hussam-i-am marked this pull request as ready for review December 18, 2024 21:31
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerDecember 18, 2024 21:31
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5438 December 18, 2024 21:31 Inactive
@github-actionsgithub-actionsBot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/355840

@hussam-i-am

hussam-i-am commented Dec 18, 2024

Copy link
Copy Markdown
Contributor

ab12bde adds the missing sx prop that was not being passed in for vertical layouts, which resolves the input not getting the full width styling that was being passed in on the "Create new issue" page.

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

RL looks good 👍

@jonrohan
jonrohan added this pull request to the merge queueDec 19, 2024
Merged via the queue into main with commit 4776d13Dec 19, 2024
@jonrohan
jonrohan deleted the refactor/update-form-control-to-css-modules-3 branch December 19, 2024 01:40
@primerprimerBot mentioned this pull request Dec 19, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@joshblack@hussam-i-am@langermank@jonrohan