Skip to content

fix(BaseStyles): remove unknown props from fallback when feature flag is enabled - #5545

Closed
joshblack wants to merge 8 commits into
mainfrom
fix/update-basestyles-system-props-spread
Closed

fix(BaseStyles): remove unknown props from fallback when feature flag is enabled#5545
joshblack wants to merge 8 commits into
mainfrom
fix/update-basestyles-system-props-spread

Conversation

@joshblack

@joshblackjoshblack commented Jan 14, 2025

Copy link
Copy Markdown
Member

Update how we pass along system props in BaseStyles when the CSS Modules flag is enabled. Currently, when the flag is enabled we are still passing along color, fontFamily, and lineHeight to the underlying element. This change updates the component to use CSS Custom Properties as a bridge between what is provided as props and what is applied to the element in CSS.

Changelog

New

Changed

  • Update BaseStyles to set common styles using CSS Custom Properties instead of as attributes on the element

Removed

Rollout strategy

  • Patch release

@changeset-bot

changeset-botBot commented Jan 14, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e40694c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jan 14, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Jan 14, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.55 KB (+0.09% 🔺)
packages/react/dist/browser.umd.js104.92 KB (+0.15% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5545 January 15, 2025 16:28 Inactive
@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 15, 2025
@joshblack
joshblack marked this pull request as ready for review January 15, 2025 16:29
@joshblack
joshblack requested a review from a team as a code ownerJanuary 15, 2025 16:29
@joshblackjoshblack removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 15, 2025
@primer
primerBot requested a review from a team as a code ownerJanuary 15, 2025 19:54
@primer
primerBot requested a review from tbenningJanuary 15, 2025 19:54
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5545 January 15, 2025 19:57 Inactive
@joshblack

Copy link
Copy Markdown
MemberAuthor

Trying to look into what's going on with the snapshot changes 🤔

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5545 January 22, 2025 19:11 Inactive
@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 22, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 22, 2025
@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 22, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5545 January 22, 2025 20:09 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 22, 2025
@joshblack

Copy link
Copy Markdown
MemberAuthor

Closing in favor of: #5606

@joshblack
joshblack deleted the fix/update-basestyles-system-props-spread branch January 27, 2025 19:42
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@francinelucca