Skip to content

fix(BaseStyles): remove unknown props from fallback when feature flag is enabled - #5606

Merged
joshblack merged 15 commits into
mainfrom
fix/update-base-styles-behind-ff
Jan 28, 2025
Merged

fix(BaseStyles): remove unknown props from fallback when feature flag is enabled#5606
joshblack merged 15 commits into
mainfrom
fix/update-base-styles-behind-ff

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update how we pass along system props in BaseStyles when the CSS Modules flag is enabled. Currently, when the flag is enabled we are still passing along color, fontFamily, and lineHeight to the underlying element. This change updates the component to use CSS Custom Properties as a bridge between what is provided as props and what is applied to the element in CSS.

Changelog

New

Changed

  • Update BaseStyles to set common styles using CSS Custom Properties instead of as attributes on the element
  • Update FeatureFlags to merge scope with parent context value

Removed

Rollout strategy

  • Patch release

@joshblack
joshblack requested a review from a team as a code ownerJanuary 27, 2025 19:42
@changeset-bot

changeset-botBot commented Jan 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d9be714

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jan 27, 2025
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 27, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Jan 27, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.39 KB (-0.06% 🔽)
packages/react/dist/browser.umd.js104.91 KB (+0.08% 🔺)

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5606 January 27, 2025 20:11 Inactive
@primer
primerBot requested a review from a team as a code ownerJanuary 27, 2025 20:21
@primer
primerBot requested a review from emilybrickJanuary 27, 2025 20:21
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5606 January 27, 2025 20:23 Inactive
@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 27, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5606 January 27, 2025 22:05 Inactive

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Noted the weight of the item titles changed in the VRT updates, intended?

@github-actions
github-actionsBot requested a deployment to storybook-preview-5606 January 28, 2025 17:51 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5606 January 28, 2025 18:00 Inactive
@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 28, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5606 January 28, 2025 18:20 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 28, 2025
@joshblack

Copy link
Copy Markdown
MemberAuthor

@francinelucca thanks for calling that out! Went to double-check and it seems like yes 👍 The current snapshots do not seem to be using the modern action list and it's probably because of the same bug with feature flags that I ran into 🤔

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/359710

@joshblack
joshblack added this pull request to the merge queueJan 28, 2025
Merged via the queue into main with commit f291e81Jan 28, 2025
@joshblack
joshblack deleted the fix/update-base-styles-behind-ff branch January 28, 2025 19:36
@primerprimerBot mentioned this pull request Jan 28, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@francinelucca