Uh oh!
There was an error while loading. Please reload this page.
[Accessibility][Storybook] Add aria-labels to the multiple item progress bar + enforce aria-label in progressbar.item - #5559
Conversation
🦋 Changeset detectedLatest commit: a489d3a The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
Uh oh!
There was an error while loading. Please reload this page.
khiga8
left a comment
There was a problem hiding this comment.
Makes sense!
- Do you think that
aria-labelshould be required onProgressBar.Item? - I see this was an Axe violation. I thought all Primer React storybook examples have Axe checks...maybe this doesn't'? do you know why this wasn't flagged?
kendallgassner
commented
Jan 21, 2025
@khiga8 I like this, good catch.
Its failing on storybook prod. 🤔 Do we not have an a CI check? |
khiga8
commented
Jan 21, 2025
There's some kind of CI axe checks running, but I'm not sure if it's running on this story... |
fcda31a to
328b979Compare👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks! |
kendallgassner
commented
Jan 21, 2025
Uh oh!
There was an error while loading. Please reload this page.
khiga8
left a comment
There was a problem hiding this comment.
Nice! Thanks for following up and investigating if/how to require the aria-label prop! Good call pivoting to a dev-only console warning! :)
do you think this component would benefit from more comprehensive improvements (re: role="meter" vs role="progressbar") at a future time, or do you think it's at a good place now?
Uh oh!
There was an error while loading. Please reload this page.
Uh oh!
There was an error while loading. Please reload this page.
kendallgassner
commented
Jan 24, 2025
Yes! I think it would be beneficial to investigate splitting up this component so that when progress is not dynamically changing we use role="meter" |
ee98f8f to
b3b2235Compare248d844 to
a413163CompareTylerJDev
commented
Jan 27, 2025
@kendallgassner do we need https://github.com/github/github/pull/358882 shipped before this PR is merged, or is it okay to merge now? I haven't updated the integration test PR, so I'm not sure if the CI is still failing or not with the new changes 🤔 |
kendallgassner
commented
Jan 27, 2025
@TylerJDev I changed this PR to console.warn users (only in dev mode) to add an aria-label so we can merge before https://github.com/github/github/pull/358882 is merged |
Closeshttps://github.com/github/accessibility-audits/issues/10134
In this PR I add an aria-label to each role='progressbar'. I also have updated the ProgressBar.item props to enforce that aria-labels are always added
Changelog
New
Changed
Removed
Rollout strategy
I am updating a storybook story.
Testing & Reviewing
Tested with a screen reader.
Merge checklist