Skip to content

Reintroduce Pagination algorithm enhancements - #5590

Merged
francinelucca merged 11 commits into
mainfrom
revert-5560-revert-5504-hectahertz/pagination-algorithm-enhancements
Jan 30, 2025
Merged

Reintroduce Pagination algorithm enhancements#5590
francinelucca merged 11 commits into
mainfrom
revert-5560-revert-5504-hectahertz/pagination-algorithm-enhancements

Conversation

@hectahertz

Copy link
Copy Markdown
Contributor

Reverts #5560

The original cause of the failing tests was that changing aria-disabled with aria-hidden triggered the contrast rules.

Here we add aria-disabled back.

@hectahertz
hectahertz requested review from a team as code ownersJanuary 23, 2025 15:39
@changeset-bot

changeset-botBot commented Jan 23, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9955465

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 23, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@hectahertzhectahertz self-assigned this Jan 23, 2025
@francineluccafrancinelucca changed the title Revert "Revert "Pagination algorithm enhancements""Reintroduce Pagination algorithm enhancementsJan 23, 2025
@github-actions

github-actionsBot commented Jan 23, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.42 KB (-0.14% 🔽)
packages/react/dist/browser.umd.js104.83 KB (-0.05% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5590 January 23, 2025 15:48 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jan 23, 2025
@hectahertz

Copy link
Copy Markdown
ContributorAuthor

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5590 January 29, 2025 04:52 Inactive
@hectahertz
hectahertz deleted the revert-5560-revert-5504-hectahertz/pagination-algorithm-enhancements branch January 30, 2025 15:36

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/360135

@github-actions
github-actionsBot requested a deployment to storybook-preview-5590 January 30, 2025 16:08 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5590 January 30, 2025 16:23 Inactive
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Jan 30, 2025
@francinelucca
francinelucca added this pull request to the merge queueJan 30, 2025
Merged via the queue into main with commit d045ae4Jan 30, 2025
@primerprimerBot mentioned this pull request Jan 30, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@hectahertz@francinelucca