Skip to content

Make Dialog footer scrollable on very short viewports - #5629

Merged
hectahertz merged 17 commits into
mainfrom
hectahertz/dialog-footer-wrapping
Mar 28, 2025
Merged

Make Dialog footer scrollable on very short viewports#5629
hectahertz merged 17 commits into
mainfrom
hectahertz/dialog-footer-wrapping

Conversation

@hectahertz

Copy link
Copy Markdown
Contributor

Part of the work to improve reflow of the component. This helps to make all the footer buttons accessible on small screens and gives some extra breathing room to the body.

Changelog

Changed

Dialog footer becomes scrollable on very short viewports.

Before

Screen.Recording.2025-01-30.at.20.46.28.mov

After

Screen.Recording.2025-01-30.at.20.46.08.mov

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

The CSS changes seem a bit esoteric, but are due to overflow: scroll not being compatible with justify-content: flex-end. flex-direction: row-reverse; flex-wrap: wrap-reverse; plus reversing the items is visually equivalent but compatible. The focus order keeps being the same.

Merge checklist

@hectahertz
hectahertz requested a review from a team as a code ownerJanuary 30, 2025 19:52
@changeset-bot

changeset-botBot commented Jan 30, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 84c7ac5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 30, 2025
@github-actions

github-actionsBot commented Jan 30, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.35 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js105.69 KB (-0.09% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 February 7, 2025 12:50 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 February 7, 2025 18:53 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 7, 2025
@primer
primerBot requested a review from a team as a code ownerFebruary 7, 2025 22:02
@primer
primerBot requested a review from emilybrickFebruary 7, 2025 22:02
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 7, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 February 7, 2025 22:15 Inactive
Comment threadpackages/react/src/Dialog/Dialog.module.css Outdated

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Left a few more comments, curious on what you think! Looks almost good to go 🚀

Comment threadpackages/react/src/Dialog/Dialog.tsx Outdated
Comment threadpackages/react/src/Dialog/Dialog.tsx Outdated
@lesliecdubs

Copy link
Copy Markdown
Member

Ping @hectahertz, would be awesome to get this one over the line before you fully switch focus to the next epic 🙇🏻‍♀️

@hectahertz
hectahertzforce-pushed the hectahertz/dialog-footer-wrapping branch from c0bee5e to e62c9a9CompareMarch 20, 2025 18:10
@hectahertz
hectahertzforce-pushed the hectahertz/dialog-footer-wrapping branch from e62c9a9 to 9e64451CompareMarch 20, 2025 18:12
@github-actions
github-actionsBot requested a deployment to storybook-preview-5629 March 24, 2025 16:26 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 March 24, 2025 16:46 Inactive
@hectahertzhectahertz added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 24, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 24, 2025
@hectahertzhectahertz added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 March 27, 2025 09:56 Inactive
@hectahertzhectahertz added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 28, 2025
@hectahertz
hectahertz enabled auto-merge March 28, 2025 17:59
@github-actions
github-actionsBot requested a deployment to storybook-preview-5629 March 28, 2025 18:03 Abandoned
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 28, 2025
@github-actions
github-actionsBot requested a deployment to storybook-preview-5629 March 28, 2025 18:14 Abandoned
@hectahertz
hectahertz added this pull request to the merge queueMar 28, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5629 March 28, 2025 18:26 Inactive
Merged via the queue into main with commit b7346b2Mar 28, 2025
@hectahertz
hectahertz deleted the hectahertz/dialog-footer-wrapping branch March 28, 2025 18:36
@primerprimerBot mentioned this pull request Mar 28, 2025
francinelucca added a commit that referenced this pull request Mar 29, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@hectahertz@lesliecdubs@joshblack@TylerJDev