Skip to content

SubNav: Add aria-current to SubNav - #5661

Merged
TylerJDev merged 4 commits into
mainfrom
tylerjdev/add-aria-current-to-subnav
Feb 13, 2025
Merged

SubNav: Add aria-current to SubNav#5661
TylerJDev merged 4 commits into
mainfrom
tylerjdev/add-aria-current-to-subnav

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Feb 5, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3433

Context: https://github.com/github/accessibility-audits/issues/10093

Changelog

New

  • Add aria-current to SubNav.Link

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@TylerJDev
TylerJDev requested a review from a team as a code ownerFebruary 5, 2025 14:53
@changeset-bot

changeset-botBot commented Feb 5, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 657fc31

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Feb 5, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Feb 5, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.69 KB (+0.12% 🔺)
packages/react/dist/browser.umd.js106.01 KB (-0.03% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5661 February 5, 2025 15:25 Inactive
@TylerJDevTylerJDev changed the title Add aria-current to SubNavSubNav: Add aria-current to SubNavFeb 5, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5661 February 5, 2025 15:39 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueFeb 13, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 13, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5661 February 13, 2025 18:34 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5661 February 13, 2025 18:43 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueFeb 13, 2025
Merged via the queue into main with commit 1262f5aFeb 13, 2025
@TylerJDev
TylerJDev deleted the tylerjdev/add-aria-current-to-subnav branch February 13, 2025 18:55
@primerprimerBot mentioned this pull request Feb 13, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
* Add `aria-current` to `SubNav`
* Add changeset
* Update snapshot
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack