Skip to content

bugfix(SelectPanel): Set border-top-color on selectpanel footer - #5712

Merged
joshblack merged 2 commits into
mainfrom
set_border_properly_on_selectpanel_footer
Feb 19, 2025
Merged

bugfix(SelectPanel): Set border-top-color on selectpanel footer#5712
joshblack merged 2 commits into
mainfrom
set_border_properly_on_selectpanel_footer

Conversation

@jonrohan

@jonrohanjonrohan commented Feb 19, 2025

Copy link
Copy Markdown
Member

@langermank noticed an issue with the selectpanel footer that had a dark line.

image

Changelog

New

Changed

Set border-top-color for SelectPanel footer

Removed

Rollout strategy

The issue is that we were setting border-top: and border-color but the border-top property is taking precedence and overriding the color.

image

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerFebruary 19, 2025 22:44
@changeset-bot

changeset-botBot commented Feb 19, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fbcd773

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Feb 19, 2025
@github-actions

github-actionsBot commented Feb 19, 2025

Copy link
Copy Markdown
Contributor

Uh oh! @jonrohan, at least one image you shared is missing helpful alt text. Check your pull request body to fix the following violations:

  • Images should have meaningful alternative text (alt text) at line 3
  • Images should have meaningful alternative text (alt text) at line 27

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

🤖 Beep boop! This comment was added automatically by github/accessibility-alt-text-bot.

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Feb 19, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice 🚀

@github-actions

github-actionsBot commented Feb 19, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.74 KB (0%)
packages/react/dist/browser.umd.js106.06 KB (0%)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5712 February 19, 2025 22:48 Abandoned
@jonrohan
jonrohan added this pull request to the merge queueFeb 19, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5712 February 19, 2025 23:01 Inactive
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 19, 2025
@joshblack
joshblack added this pull request to the merge queueFeb 19, 2025
Merged via the queue into main with commit 22508b7Feb 19, 2025
@joshblack
joshblack deleted the set_border_properly_on_selectpanel_footer branch February 19, 2025 23:23
@primerprimerBot mentioned this pull request Feb 19, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
* Set border-top-color on selectpanel footer
* Create clean-planets-sleep.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@joshblack@langermank