Skip to content

Radio: Remove aria-required and aria-invalid - #5760

Merged
TylerJDev merged 5 commits into
mainfrom
tylerjdev/remove-lint-disables
Mar 14, 2025
Merged

Radio: Remove aria-required and aria-invalid#5760
TylerJDev merged 5 commits into
mainfrom
tylerjdev/remove-lint-disables

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Mar 11, 2025

Copy link
Copy Markdown
Member

Context https://github.com/github/primer/issues/3726

Removes aria-required and aria-invalid from the Radio component, as they are not required on the component. This change allows us to remove a lint disable for a11y-role-supports-aria-props.

Changelog

Removed

  • Removes aria-invalid and aria-required from Radio

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 11, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bc31fb5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 11, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 11, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.97 KB (-0.07% 🔽)
packages/react/dist/browser.umd.js106.42 KB (+0.08% 🔺)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5760 March 11, 2025 14:45 Abandoned
@github-actions
github-actionsBot requested a deployment to storybook-preview-5760 March 11, 2025 14:52 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5760 March 11, 2025 15:04 Inactive
/**
* Only used to inform ARIA attributes. Individual radio inputs do not have validation styles.
*/
validationStatus?: FormValidationStatus

@TylerJDevTylerJDevMar 11, 2025

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We don't advertise this prop in our docs, and it doesn't have any usage outside of the aria-* props which are removed in this PR. The prop itself didn't add any functionality and I don't think this would need to be a major release. There's only one place (outside of Dotcom) that appears to utilize this prop.

Any thoughts on this?

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@TylerJDev if you feel like we can safely roll this out then I'm okay with that 👍 I think it would fall under a breaking change but I get if it doesn't feel worth it to migrate something we don't advertise and doesn't have usage.

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Definitely! I think this is one of those things that didn't really do anything, and was never adopted anywhere, so we can probably get away with just removing it 😄

@TylerJDev
TylerJDev marked this pull request as ready for review March 11, 2025 18:35
@TylerJDev
TylerJDev requested a review from a team as a code ownerMarch 11, 2025 18:35
@TylerJDevTylerJDev mentioned this pull request Mar 11, 2025
13 tasks
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/366482

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 11, 2025
@TylerJDev
TylerJDev added this pull request to the merge queueMar 13, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Mar 13, 2025
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 13, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5760 March 13, 2025 19:36 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueMar 14, 2025
Merged via the queue into main with commit 8b492cfMar 14, 2025
@TylerJDev
TylerJDev deleted the tylerjdev/remove-lint-disables branch March 14, 2025 13:44
@primerprimerBot mentioned this pull request Mar 13, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack