Skip to content

[CSS Module Migration] Remove feature flag for Underline Panel - #5836

Merged
jamieshark merged 6 commits into
mainfrom
css_modules_remove_flag_underlinepanels
Apr 2, 2025
Merged

[CSS Module Migration] Remove feature flag for Underline Panel#5836
jamieshark merged 6 commits into
mainfrom
css_modules_remove_flag_underlinepanels

Conversation

@jamieshark

@jamiesharkjamieshark commented Mar 28, 2025

Copy link
Copy Markdown
Contributor

This PR removes the CSS modules feature flag from the FormControl component. The component UnderlinePanel is used 0 times in dotcom.

Changelog

New

Changed

Removed

Feature flag for Underline Panel and associated components

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 28, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0bad285

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 28, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 28, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.54 KB (-0.46% 🔽)
packages/react/dist/browser.umd.js105.82 KB (-0.52% 🔽)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5836 April 1, 2025 21:25 Abandoned
@jamieshark
jamieshark marked this pull request as ready for review April 1, 2025 21:33
@jamieshark
jamieshark requested a review from a team as a code ownerApril 1, 2025 21:33
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 1, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5836 April 1, 2025 21:37 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 1, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5836 April 1, 2025 21:47 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 1, 2025

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM barring integration failures

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5836 April 2, 2025 18:08 Inactive
@jamiesharkjamieshark added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/370942

@primer
primerBot requested a review from a team as a code ownerApril 2, 2025 19:04
@primer
primerBot requested a review from mperrottiApril 2, 2025 19:04
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5836 April 2, 2025 19:18 Inactive
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Apr 2, 2025
@jamieshark
jamieshark added this pull request to the merge queueApr 2, 2025
Merged via the queue into main with commit c177e5eApr 2, 2025
@jamieshark
jamieshark deleted the css_modules_remove_flag_underlinepanels branch April 2, 2025 21:06
@primerprimerBot mentioned this pull request Apr 2, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jamieshark@francinelucca@langermank