Skip to content

Simplify arguments for toggle component helper function - #5861

Merged
jonrohan merged 14 commits into
mainfrom
jshark/toggle-component-fix
Apr 3, 2025
Merged

Simplify arguments for toggle component helper function#5861
jonrohan merged 14 commits into
mainfrom
jshark/toggle-component-fix

Conversation

@jamieshark

@jamiesharkjamieshark commented Apr 1, 2025

Copy link
Copy Markdown
Contributor

Changelog

New

Changed

Removed

  • Removes sx argument in toggleSxComponent

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jamieshark
jamieshark requested a review from a team as a code ownerApril 1, 2025 21:26
@changeset-bot

changeset-botBot commented Apr 1, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bb94f5f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Apr 1, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Apr 1, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.98 KB (-0.02% 🔽)
packages/react/dist/browser.umd.js105.35 KB (-0.04% 🔽)

@jamieshark

Copy link
Copy Markdown
ContributorAuthor

merge #5836 before this PR

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5861 April 1, 2025 21:39 Inactive
@jonrohan
jonrohan changed the base branch from main to css_modules_remove_flag_underlinepanelsApril 2, 2025 03:24
@jamiesharkjamieshark self-assigned this Apr 2, 2025
Base automatically changed from css_modules_remove_flag_underlinepanels to mainApril 2, 2025 21:06
@jamieshark
jamieshark requested a review from a team as a code ownerApril 2, 2025 21:06
@jamieshark
jamieshark requested a review from tbenningApril 2, 2025 21:06
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5861 April 2, 2025 21:10 Inactive
@jamieshark
jamieshark marked this pull request as draft April 2, 2025 22:06
@github-actions
github-actionsBot requested a deployment to storybook-preview-5861 April 2, 2025 22:11 Abandoned
@github-actions
github-actionsBot requested a deployment to storybook-preview-5861 April 3, 2025 15:22 Abandoned
@jonrohan
jonrohan added this pull request to the merge queueApr 3, 2025
Merged via the queue into main with commit ecb449fApr 3, 2025
@jonrohan
jonrohan deleted the jshark/toggle-component-fix branch April 3, 2025 17:08
@primerprimerBot mentioned this pull request Apr 3, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jamieshark@jonrohan@joshblack