Skip to content

SelectPanelV2: Deprecate SelectPanelV2 - #5895

Closed
TylerJDev wants to merge 8 commits into
mainfrom
deprecate-selectpanel2
Closed

SelectPanelV2: Deprecate SelectPanelV2#5895
TylerJDev wants to merge 8 commits into
mainfrom
deprecate-selectpanel2

Conversation

@TylerJDev

@TylerJDev TylerJDev commented Apr 7, 2025

Copy link
Copy Markdown
Member

Closes https://github.com/github/primer/issues/4786

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-bot Bot commented Apr 7, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c88c1b0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@primer/react Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 7, 2025
@github-actions

github-actions Bot commented Apr 7, 2025

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actions Bot commented Apr 7, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

Path Size
packages/react/dist/browser.esm.js 104.47 KB (0%)
packages/react/dist/browser.umd.js 104.91 KB (0%)

@github-actions
github-actions Bot requested a deployment to storybook-preview-5895 April 7, 2025 13:57 Abandoned
@github-actions
github-actions Bot requested a deployment to storybook-preview-5895 April 7, 2025 14:03 Abandoned
@github-actions
github-actions Bot requested a deployment to storybook-preview-5895 April 7, 2025 15:25 Abandoned
@github-actions
github-actions Bot temporarily deployed to storybook-preview-5895 April 7, 2025 15:38 Inactive
@github-actions
github-actions Bot temporarily deployed to storybook-preview-5895 April 7, 2025 16:08 Inactive
@TylerJDev TylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 7, 2025
@github-actions github-actions Bot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 7, 2025
@github-actions
github-actions Bot temporarily deployed to storybook-preview-5895 April 7, 2025 16:35 Inactive
@TylerJDev TylerJDev changed the title SelectPanel2: Deprecate SelectPanel2 SelectPanelV2: Deprecate SelectPanelV2 Apr 7, 2025
@TylerJDev

Copy link
Copy Markdown
Member Author

Continued in #5906.

@TylerJDev TylerJDev closed this Apr 8, 2025
@primer
primer Bot deleted the deprecate-selectpanel2 branch June 30, 2026 18:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm staff Author is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant