Skip to content

SelectPanelV2: Add Deprecated status to SelectPanelV2 docs, stories - #5906

Merged
TylerJDev merged 2 commits into
mainfrom
deprecate-selectpanel-v2
Apr 9, 2025
Merged

SelectPanelV2: Add Deprecated status to SelectPanelV2 docs, stories#5906
TylerJDev merged 2 commits into
mainfrom
deprecate-selectpanel-v2

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Apr 8, 2025

Copy link
Copy Markdown
Member

Relates to https://github.com/github/primer/issues/4786

Adds deprecated status to PRC experimental SelectPanel

Changelog

Changed

  • Adds deprecated status to PRC experimental SelectPanel, moves SelectPanel stories to deprecated folder

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Apr 8, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a11956b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Apr 8, 2025
@TylerJDevTylerJDev mentioned this pull request Apr 8, 2025
13 tasks
@github-actions

github-actionsBot commented Apr 8, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.35 KB (0%)
packages/react/dist/browser.umd.js104.74 KB (0%)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5906 April 8, 2025 19:01 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5906 April 8, 2025 19:15 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review April 8, 2025 23:27
@TylerJDev
TylerJDev requested a review from a team as a code ownerApril 8, 2025 23:27

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@TylerJDev
TylerJDev added this pull request to the merge queueApr 9, 2025
Merged via the queue into main with commit 8fa12eaApr 9, 2025
@TylerJDev
TylerJDev deleted the deprecate-selectpanel-v2 branch April 9, 2025 12:25
@primerprimerBot mentioned this pull request Apr 9, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@francinelucca