Skip to content

chore(ActionList, NavList): Remove the CSS modules feature flag from ActionList and related components - #6090

Merged
jonrohan merged 18 commits into
mainfrom
css_modules_remove_flag/actionlist
Jun 11, 2025
Merged

chore(ActionList, NavList): Remove the CSS modules feature flag from ActionList and related components#6090
jonrohan merged 18 commits into
mainfrom
css_modules_remove_flag/actionlist

Conversation

@jonrohan

@jonrohanjonrohan commented May 20, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/5173

Changelog

New

Changed

Removed

Remove the CSS modules feature flag from ActionList and related components

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented May 20, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7aebf52

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label May 20, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label May 20, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6090 May 20, 2025 21:36 Inactive
@github-actions

github-actionsBot commented May 20, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js91.93 KB (-2.74% 🔽)
packages/react/dist/browser.umd.js92.03 KB (-2.93% 🔽)

@github-actions
github-actionsBot requested a deployment to storybook-preview-6090 May 21, 2025 00:11 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6090 May 21, 2025 00:25 Inactive
className?: string
groupId?: string
renderItem?: (item: React.FC<React.PropsWithChildren<MenuItemProps>>) => React.ReactNode
handleAddItem?: (item: React.FC<React.PropsWithChildren<MenuItemProps>>) => void

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Wanted to highlight this as something I'm unsure of? @joshblack I had to add these here because of the props I added on the item to satisfy the type errors

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels May 21, 2025
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Jun 4, 2025
@hussam-i-am
hussam-i-am marked this pull request as ready for review June 4, 2025 14:28
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerJune 4, 2025 14:28
@hussam-i-am

Copy link
Copy Markdown
Contributor

Integration tests are now green, but the realease will need the changes in https://github.com/github/github/pull/381456 for them to pass.

@jonrohan
jonrohan added this pull request to the merge queueJun 11, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Jun 11, 2025
@jonrohan
jonrohan added this pull request to the merge queueJun 11, 2025
@jonrohan
jonrohan removed this pull request from the merge queue due to a manual request Jun 11, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jun 11, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6090 June 11, 2025 18:30 Inactive
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jun 11, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@jonrohan
jonrohan enabled auto-merge June 11, 2025 18:36
@jonrohan
jonrohan added this pull request to the merge queueJun 11, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6090 June 11, 2025 18:51 Inactive
Merged via the queue into main with commit 7bbdcabJun 11, 2025
@jonrohan
jonrohan deleted the css_modules_remove_flag/actionlist branch June 11, 2025 19:00
@primerprimerBot mentioned this pull request Jun 11, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@hussam-i-am@francinelucca