Skip to content

Remove support for sx from FormControl component - #6681

Merged
francinelucca merged 29 commits into
mainfrom
llastflowers/5794/no-sx-FormControl
Sep 29, 2025
Merged

Remove support for sx from FormControl component#6681
francinelucca merged 29 commits into
mainfrom
llastflowers/5794/no-sx-FormControl

Conversation

@llastflowers

@llastflowersllastflowers commented Aug 27, 2025

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/5794
Closeshttps://github.com/github/primer/issues/5909

Current sx usage: 160

⚠️ Need to separately update stories that depend on this component [SelectPanel.examples.stories.tsx]

Changelog

Removed

Remove support for sx from the FormControl component, and associated stories, docs, and tests

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

We are using the @primer/styled-react component to mitigate removing sx usage. We will make sure usage is updated upstream before merging.

Merge checklist

@llastflowersllastflowers self-assigned this Aug 27, 2025
@changeset-bot

changeset-botBot commented Aug 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e81d299

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
NameType
@primer/reactMajor
@primer/styled-reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Aug 27, 2025
@github-actions

github-actionsBot commented Aug 27, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js89.58 KB (-0.03% 🔽)
packages/react/dist/browser.umd.js89.58 KB (-0.12% 🔽)

@github-actions
github-actionsBot requested a deployment to storybook-preview-6681 August 27, 2025 19:17 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6681 August 27, 2025 19:25 Inactive
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 26, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-6681 September 26, 2025 20:57 Inactive
@github-actionsgithub-actionsBot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 26, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6681 September 26, 2025 21:05 Inactive
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 26, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-6681 September 26, 2025 21:31 Inactive
@github-actionsgithub-actionsBot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 26, 2025
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github-ui! Your integration PR is ready: https://github.com/github/github-ui/pull/3429

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh labels Sep 26, 2025
@primer-integration

Copy link
Copy Markdown

🟢 ci completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Sep 26, 2025
@francinelucca
francinelucca added this pull request to the merge queueSep 29, 2025
Merged via the queue into main with commit 158703bSep 29, 2025
43 checks passed
@francinelucca
francinelucca deleted the llastflowers/5794/no-sx-FormControl branch September 29, 2025 19:25
@primerprimerBot mentioned this pull request Sep 29, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@llastflowers@francinelucca@liuliu-dev