Uh oh!
There was an error while loading. Please reload this page.
feat(FormControl): remove sx - #6951
Conversation
🦋 Changeset detectedLatest commit: 0df4777 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks! |
👋 Hi from github/github-ui! Your integration PR is ready: https://github.com/github/github-ui/pull/3933 |
…eact into feat/remove-sx-from-formcontrol
👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks! |
1 similar comment
👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks! |
👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks! |
🔴 ci completed with status |
francinelucca
commented
Oct 11, 2025
Failure is on main branch, merging https://github.com/github/github-ui/pull/4409 |
Reintroduces #6681 without the subcomponent wrappers since they won't play well with slots