Skip to content

Remove sx from the TextInput component - #6717

Closed
liuliu-dev wants to merge 10 commits into
mainfrom
liuliu/remove-sx-from-textinput
Closed

Remove sx from the TextInput component#6717
liuliu-dev wants to merge 10 commits into
mainfrom
liuliu/remove-sx-from-textinput

Conversation

@liuliu-dev

Copy link
Copy Markdown
Contributor

Remove support for sx from the TextInput
Closes#5828

sx usage

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

@changeset-bot

changeset-botBot commented Sep 3, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 8fb865c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
NameType
@primer/reactMajor
@primer/styled-reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 3, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions
github-actionsBot requested a deployment to storybook-preview-6717 September 3, 2025 22:40 Abandoned
@github-actions

github-actionsBot commented Sep 3, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js88.57 KB (-0.09% 🔽)
packages/react/dist/browser.umd.js88.75 KB (+0.03% 🔺)

@liuliu-devliuliu-dev changed the title Remove sx rom the TextInput componentRemove sx from the TextInput componentSep 3, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6717 September 3, 2025 23:06 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6717 September 11, 2025 21:22 Inactive
@liuliu-dev

Copy link
Copy Markdown
ContributorAuthor

closing as it got handled in this PR: #6666

@jonrohan
jonrohan deleted the liuliu/remove-sx-from-textinput branch October 30, 2025 21:01
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@liuliu-dev