Skip to content

ActionList: Fix icon and description hover color for danger variant - #3505

Merged
siddharthkp merged 11 commits into
mainfrom
fixes-3044
Jul 12, 2023
Merged

ActionList: Fix icon and description hover color for danger variant#3505
siddharthkp merged 11 commits into
mainfrom
fixes-3044

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Jul 10, 2023

Copy link
Copy Markdown
Member

@siddharthkp
siddharthkp requested a review from a teamJuly 10, 2023 10:55
@siddharthkpsiddharthkp self-assigned this Jul 10, 2023
@changeset-bot

changeset-botBot commented Jul 10, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4068169

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

This comment was marked as resolved.

@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 10, 2023
@siddharthkp
siddharthkp requested review from joshblack and removed request for josepmartinsJuly 10, 2023 10:58
@github-actions

github-actionsBot commented Jul 10, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js102.37 KB (+0.06% 🔺)
dist/browser.umd.js102.92 KB (+0.05% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3505 July 10, 2023 11:03 Inactive
@siddharthkp
siddharthkp temporarily deployed to github-pages July 10, 2023 11:03 — with GitHub Actions Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 10, 2023
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3505 July 10, 2023 11:05 Inactive
@siddharthkp
siddharthkp changed the base branch from main to fixtures-3044July 10, 2023 11:09
@primer
primerBot temporarily deployed to github-pages July 10, 2023 11:09 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3505 July 10, 2023 11:10 Inactive
Base automatically changed from fixtures-3044 to mainJuly 10, 2023 15:54
@siddharthkpsiddharthkp added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 11, 2023
@siddharthkp
siddharthkp temporarily deployed to github-pages July 11, 2023 14:58 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3505 July 11, 2023 14:59 Inactive
@siddharthkpsiddharthkp added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Jul 11, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jul 11, 2023
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3505 July 11, 2023 16:44 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Love to see a clean diff :feels-good:

@siddharthkp
siddharthkp added this pull request to the merge queueJul 12, 2023
Merged via the queue into main with commit c3f1e3cJul 12, 2023
@siddharthkp
siddharthkp deleted the fixes-3044 branch July 12, 2023 09:48
This was referenced Jul 12, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@joshblack@langermank