Skip to content

ButtonGroup: Fix button and icon button styling when tooltips are used on them (#4210) - #4251

Closed
broccolinisoup wants to merge 11 commits into
mainfrom
re-introduce-button-group-tooltip-fix
Closed

ButtonGroup: Fix button and icon button styling when tooltips are used on them (#4210)#4251
broccolinisoup wants to merge 11 commits into
mainfrom
re-introduce-button-group-tooltip-fix

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Feb 11, 2024

Copy link
Copy Markdown
Member

Closes#4129

Reintroducing #4210 since it is reverted due to random VRT tests failings - I removed the threshold setting in Playwright since the default one is less strict.

Changelog

New

Changed

Updated Button group CSS to make sure the icon buttons and buttons are rendered as expected when they have tooltips (both v1 and v2) on them

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

…d on them (#4210)
* fix button group styling for tooltip and tooltip v2
* Add changeset
* take the a tag into account when styling
* add e2e tests for the buttons groups that use tooltip
* test(vrt): update snapshots
---------
Co-authored-by: broccolinisoup <broccolinisoup@users.noreply.github.com>
@changeset-bot

changeset-botBot commented Feb 11, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f440bec

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 11, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.64 KB (+0.07% 🔺)
packages/react/dist/browser.umd.js114.34 KB (+0.11% 🔺)

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Seems good from a quick pass, is it still cutting it off by 1px like you were noticing before? 👀

@broccolinisoup
broccolinisoup marked this pull request as ready for review March 2, 2024 00:59
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4251 March 2, 2024 01:02 Inactive
@broccolinisoupbroccolinisoup added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Mar 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4251 March 4, 2024 00:06 Inactive
@broccolinisoup
broccolinisoupforce-pushed the re-introduce-button-group-tooltip-fix branch from 4a24bf5 to acfa559CompareMarch 4, 2024 00:54
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4251 March 4, 2024 00:58 Inactive
@broccolinisoupbroccolinisoup added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Mar 4, 2024
@broccolinisoupbroccolinisoup removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 4, 2024
@broccolinisoupbroccolinisoup added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4251 March 4, 2024 04:44 Inactive
@broccolinisoupbroccolinisoup added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Mar 4, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4251 March 4, 2024 05:49 Inactive
@broccolinisoup

Copy link
Copy Markdown
MemberAuthor

Seems good from a quick pass, is it still cutting it off by 1px like you were noticing before? 👀

I think it is all gone now, removing the strict threshold worked 👍🏻

@broccolinisoup

Copy link
Copy Markdown
MemberAuthor

closing this in favour of #4371

@broccolinisoup
broccolinisoup deleted the re-introduce-button-group-tooltip-fix branch April 5, 2024 02:47
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Using <Tooltip> within <ButtonGroup> affects styling

3 participants

@broccolinisoup@joshblack@langermank