Skip to content

Bring back #5255 "SelectPanel: Prepare for non-anchored variants" - #5262

Merged
broccolinisoup merged 3 commits into
mainfrom
revert-5255-revert-5230-selectpanel-use-anchored-position
Nov 12, 2024
Merged

Bring back #5255 "SelectPanel: Prepare for non-anchored variants"#5262
broccolinisoup merged 3 commits into
mainfrom
revert-5255-revert-5230-selectpanel-use-anchored-position

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 11, 2024

Copy link
Copy Markdown
Member

diff from PR that had to be reverted:

- if (!open) return <>{anchor}</>+ if (!open) return <LiveRegion>{anchor}</LiveRegion>

The reason I had to revert it is because i added one commit after running integration tests that was faulty 🤦

@siddharthkpsiddharthkp self-assigned this Nov 11, 2024
@changeset-bot

changeset-botBot commented Nov 11, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 080e0f6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 11, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5262 November 11, 2024 16:29 Inactive
@github-actions

github-actionsBot commented Nov 11, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.59 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js99.02 KB (+0.16% 🔺)

@primer-integration

primer-integrationBot commented Nov 11, 2024

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 11, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/350750

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Nov 11, 2024
@siddharthkp
siddharthkp marked this pull request as ready for review November 11, 2024 19:14
@siddharthkp
siddharthkp requested a review from a team as a code ownerNovember 11, 2024 19:14
@siddharthkpsiddharthkp changed the title Revert "Revert "SelectPanel: Prepare for non-anchored variants""Bring back "SelectPanel: Prepare for non-anchored variants""Nov 11, 2024
@siddharthkpsiddharthkp changed the title Bring back "SelectPanel: Prepare for non-anchored variants""Bring back #5255 "SelectPanel: Prepare for non-anchored variants"Nov 11, 2024
@broccolinisoup
broccolinisoup added this pull request to the merge queueNov 12, 2024
Merged via the queue into main with commit fca4ec9Nov 12, 2024
@broccolinisoup
broccolinisoup deleted the revert-5255-revert-5230-selectpanel-use-anchored-position branch November 12, 2024 02:14
@primerprimerBot mentioned this pull request Nov 12, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@broccolinisoup