Skip to content

chore(ButtonGroup): Remove the CSS modules feature flag from ButtonGroup - #5339

Merged
jonrohan merged 8 commits into
mainfrom
css_modules_remove_flag_buttongroup
Dec 5, 2024
Merged

chore(ButtonGroup): Remove the CSS modules feature flag from ButtonGroup#5339
jonrohan merged 8 commits into
mainfrom
css_modules_remove_flag_buttongroup

Conversation

@jonrohan

@jonrohanjonrohan commented Nov 27, 2024

Copy link
Copy Markdown
Member

Closes #

Changelog

New

Changed

Removed

Remove the CSS modules feature flag from ButtonGroup

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Nov 27, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ef4743e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 27, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Nov 27, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.93 KB (-0.14% 🔽)
packages/react/dist/browser.umd.js104.38 KB (-0.13% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5339 November 27, 2024 22:26 Inactive
@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 27, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 27, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5339 November 27, 2024 22:52 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5339 December 2, 2024 22:35 Inactive
@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 2, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 2, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/353480

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 4, 2024
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Dec 5, 2024
@jonrohan
jonrohan marked this pull request as ready for review December 5, 2024 18:11
@jonrohan
jonrohan requested review from a team as code ownersDecember 5, 2024 18:11
@jonrohan
jonrohan added this pull request to the merge queueDec 5, 2024
Merged via the queue into main with commit dc2f083Dec 5, 2024
@jonrohan
jonrohan deleted the css_modules_remove_flag_buttongroup branch December 5, 2024 18:33
@primerprimerBot mentioned this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
jonrohan added a commit that referenced this pull request Dec 10, 2024
github-merge-queueBot pushed a commit that referenced this pull request Dec 11, 2024
* Revert "chore(ButtonGroup): Remove the CSS modules feature flag from ButtonGroup (#5339)"
This reverts commit dc2f083.
* Revert "BugFix: Use first-of-type for buttongroup selector (#5343)"
This reverts commit 70005b8.
* Add changeset
* @primer/react v37.7.1
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack