Skip to content

BugFix: Use first-of-type for buttongroup selector - #5343

Merged
jonrohan merged 3 commits into
mainfrom
buttongroup_last_of_type_fix
Dec 3, 2024
Merged

BugFix: Use first-of-type for buttongroup selector#5343
jonrohan merged 3 commits into
mainfrom
buttongroup_last_of_type_fix

Conversation

@jonrohan

@jonrohanjonrohan commented Dec 3, 2024

Copy link
Copy Markdown
Member

While cleaning up the ButtonGroup feature flags #5339, I noticed a bug in the Overlay screenshots where the last-child button didn't have the correct border radius.

This is because of the hidden Tooltip component in the group of items. I updated the selector to use :is(button, a) and :first-of-type and :last-of-type pseudo-classes to fix the issue. The updated screenshots show the issue.

Changelog

New

Changed

Fix issue in ButtonGroup Overlay screenshots by updating the selector to use :is(button, a) and :first-of-type and :last-of-type pseudo-classes.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 426beb0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 3, 2024
@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 3, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.53 KB (+0.11% 🔺)
packages/react/dist/browser.umd.js103.8 KB (+0.01% 🔺)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5343 December 3, 2024 00:17 Inactive
@jonrohan
jonrohan added this pull request to the merge queueDec 3, 2024
Merged via the queue into main with commit 70005b8Dec 3, 2024
@jonrohan
jonrohan deleted the buttongroup_last_of_type_fix branch December 3, 2024 01:21
@primerprimerBot mentioned this pull request Dec 2, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
jonrohan added a commit that referenced this pull request Dec 10, 2024
github-merge-queueBot pushed a commit that referenced this pull request Dec 11, 2024
* Revert "chore(ButtonGroup): Remove the CSS modules feature flag from ButtonGroup (#5339)"
This reverts commit dc2f083.
* Revert "BugFix: Use first-of-type for buttongroup selector (#5343)"
This reverts commit 70005b8.
* Add changeset
* @primer/react v37.7.1
@jonrohanjonrohan mentioned this pull request Dec 17, 2024
13 tasks
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank