Skip to content

chore: migrate pagination component - #5331

Merged
ovflowd merged 6 commits into
nodejs:major/website-redesignfrom
ktssr:chore/migrate-pagination-component
Apr 28, 2023
Merged

chore: migrate pagination component#5331
ovflowd merged 6 commits into
nodejs:major/website-redesignfrom
ktssr:chore/migrate-pagination-component

Conversation

@ktssr

Copy link
Copy Markdown
Contributor

No description provided.

@vercel

vercelBot commented Apr 26, 2023

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

NameStatusPreviewCommentsUpdated (UTC)
nodejs-org✅ Ready (Inspect)Visit Preview💬 Add feedbackApr 28, 2023 10:28am
nodejs-org-stories✅ Ready (Inspect)Visit PreviewApr 28, 2023 10:28am

@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 26, 2023 10:47 Inactive
@ktssr
ktssrforce-pushed the chore/migrate-pagination-component branch from 4fc3856 to 68cff74CompareApril 26, 2023 10:47
@ktssr
ktssr requested a review from a team as a code ownerApril 26, 2023 10:47
@ktssr
ktssrforce-pushed the chore/migrate-pagination-component branch from 68cff74 to 4fc3856CompareApril 26, 2023 10:49
@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 26, 2023 10:51 Inactive
@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 26, 2023 12:35 Inactive
Comment threadcomponents/Learn/PreviousNextLink/__tests__/index.test.tsx Outdated
@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 26, 2023 12:36 Inactive
Comment threadcomponents/Learn/PreviousNextLink/index.module.scss
Comment threadcomponents/Learn/PreviousNextLink/index.stories.tsx
Comment threadcomponents/Learn/PreviousNextLink/index.tsx Outdated
Comment threadcomponents/Learn/PreviousNextLink/index.tsx Outdated
Comment threadcomponents/Learn/PreviousNextLink/index.tsx Outdated
@targos

Copy link
Copy Markdown
Member

Why do you want input from the releasers team?

@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 26, 2023 13:30 Inactive
@ktssr

Copy link
Copy Markdown
ContributorAuthor

Why do you want input from the releasers team?

I apologize for any confusion, I am having difficulty understanding what you are trying to convey. could you please elaborate?

@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 26, 2023 13:35 Inactive
@targos

Copy link
Copy Markdown
Member

I am in the release team and we got a notification because we were requested as reviewers:
CleanShot 2023-04-26 at 16 03 04

@ovflowd

Copy link
Copy Markdown
Member

Really odd, @targos since the CODEWONERS shouldn't mark releasers for this changeset (https://github.com/nodejs/nodejs.org/blob/main/CODEOWNERS)

Comment threadcomponents/Learn/PreviousNextLink/index.tsx Outdated
@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 26, 2023 14:23 Inactive
@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 26, 2023 14:24 Inactive

@ovflowdovflowd left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM! Thank you @ktssr for this amazing contribution!

@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 27, 2023 01:00 Inactive
@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 27, 2023 01:01 Inactive
@vercel
vercelBot temporarily deployed to Preview – nodejs-org April 28, 2023 10:27 Inactive
@vercel
vercelBot temporarily deployed to Preview – nodejs-org-stories April 28, 2023 10:28 Inactive
@ovflowd
ovflowd merged commit d27d83a into nodejs:major/website-redesignApr 28, 2023
ovflowd added a commit that referenced this pull request May 3, 2023
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
araujogui pushed a commit to araujogui/nodejs.org that referenced this pull request May 8, 2023
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
ovflowd added a commit that referenced this pull request May 14, 2023
* feat(unit-test): introduce unit test on website redesign branch (#5178
feat(unit-test): introduce unit test
* chore(minor): just a tiny design nitpick
* chore: set up storybook (#5191) (#5214)
* chore: set up testing-library jest extend (#5231)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
* chore(dependencies): updated dependencies
* feat: create section title component (#5237)
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* 📎 chore(migration): migrate banner component (#5233)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* feat: migrate AnimatedPlaceholder component (#5238
Migrate AnimatedPlaceholder component from nodejs.dev
and create a new Story.
* feat: create article alert component (#5243)
* feat: migrate blockquote component (#5259)
* feat(stylelint,storybook): fixed styleling misconfig and fixed storybooks (#5281)
* feat: create article data tag component (#5280)
* feat: migrate AuthorList component and add story 🎉 (#5277)
Co-authored-by: Michael Esteban <mickel13@gmail.com>
* chore(migration): migrate language selector component (#5266)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Teja Sai Sandeep Reddy Konala <sandeep.konala@knacksystems.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
* feat(DarkModeToggler): Migrate and add stories to theme toggler 🎉 (#5236)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
* chore: updated contributing guidelines, eslint rules and storybook templates (#5294)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
* chore: contributing quick fix of example
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
* chore: story guide and react spreading
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
* chore: remove base styles from old styling
* chore: fix storybook styles, imports, typescript config and dependencies (#5319
* chore: optimises tsconfig
* chiore: add missing dependencies
* chore: type storybook constants
* chore: styles moved styles to somewhere else
* chore: add global json type definition
* chore: i18n aria-label instead of sr-only
* chore: added open sans font family and space between imports
* chore: moved styles and fixed styles and updated banner stories
* fix: stylelint rules
* chore: updated tsconfig
* chore: fix tests
* fix: darkmodetoggle test
* chore: stories use index.stories.tsx
* chore: adopt turborepo (#5316
* chore: revert pnpm use plain npm
* fix: package.json
* chore: remove warnings and add node_env
* chore: cross-env
* fix: fix turbo pipelines
* chore: only cache certain files
* chore: turbo shouldn't care about coverage outputs
* chore: proper inputs and outputs for pipelines
* chore: do not store some outputs and updated inputs for lint
* chore: added prettier configs
* chore: remove console.info
* chore: updated inputs of all other entries
* fix(package.json) Lint command is missing slashes (#5321
fix(package.json) lint:fix missing slashes
* moved `DataTag` to `components/Api` (#5317)
Co-authored-by: vasanth9 <cheepurupalli.vasanthkumar.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
* hot-fix: dependency updates and fix dev runtime
* chore: updated start command
* migration(Layout): newFooter (#5320)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* feat: migrate EditLink component (#5271)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* feat(blog) Migrate BlogCard component (#5323)
Co-authored-by: Michael Esteban <mickel13@gmail.com>
* Issue#5307 - Add framer-motion to the dependency list (#5318)
* chore: add remote turbo cache and simplified gh actions cache (#5326)Co-authored-by: Aymen Naghmouchi <aymenadvance@gmail.com>
* chore: add remote turbo cache and simplified gh actions cache
* chore: updated cache rules
* chore: more cache rules
---------
Co-authored-by: Aymen Naghmouchi <aymenadvance@gmail.com>
* chore: fix storybook local development mode (#5335)
* chore: migrate pagination component (#5331)
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* Chore(node feat) (#5338)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* chore: migrate releases types (#5324)
* hotfix: first element no margin-top
* (website redesign) Feat(shellbox): migration (#5234)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* fix(i18n): translation key (#5347)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* chore: updated dependencies
* chore: vercel enable middleware and i18n redirection (#5300)
* feat(unit-test): introduce unit test on website redesign branch (#5178
feat(unit-test): introduce unit test
* chore: set up storybook (#5191) (#5214)
* feat(stylelint,storybook): fixed styleling misconfig and fixed storybooks (#5281)
* feat(DarkModeToggler): Migrate and add stories to theme toggler 🎉 (#5236)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
* chore: updated contributing guidelines, eslint rules and storybook templates (#5294)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
* chore: fix storybook styles, imports, typescript config and dependencies (#5319
* chore: optimises tsconfig
* chiore: add missing dependencies
* chore: type storybook constants
* chore: styles moved styles to somewhere else
* chore: add global json type definition
* chore: i18n aria-label instead of sr-only
* chore: added open sans font family and space between imports
* chore: moved styles and fixed styles and updated banner stories
* fix: stylelint rules
* chore: updated tsconfig
* chore: fix tests
* fix: darkmodetoggle test
* chore: stories use index.stories.tsx
* chore: adopt turborepo (#5316
* chore: revert pnpm use plain npm
* fix: package.json
* chore: remove warnings and add node_env
* chore: cross-env
* fix: fix turbo pipelines
* chore: only cache certain files
* chore: turbo shouldn't care about coverage outputs
* chore: proper inputs and outputs for pipelines
* chore: do not store some outputs and updated inputs for lint
* chore: added prettier configs
* chore: remove console.info
* chore: updated inputs of all other entries
* feat(stability): migrate component
* chore(snapshot): update
* Update components/Api/Stability/index.tsx
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
* Update components/Api/Stability/index.tsx
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
* Update components/Api/Stability/index.tsx
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
* fea(stability): update stories + fix
* Update components/Api/Stability/index.stories.tsx
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
* fix(i18n): update turkish language direction (#5182)
* chore: rollback CODEOWNER changes (#5183)
* Sync: merge `major/website-redesign` into `main` (#5356)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Teja Sai Sandeep Reddy Konala <sandeep.konala@knacksystems.com>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
Co-authored-by: vasanth9 <cheepurupalli.vasanthkumar.com>
Co-authored-by: Aymen Naghmouchi <aymenadvance@gmail.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Augustin Mauroy <97875033+AugustinMauroy@users.noreply.github.com>
Co-authored-by: Guilherme Araújo <guilherme.araujo@maxxidata.com>
Co-authored-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
Co-authored-by: HinataKah0 <128208841+HinataKah0@users.noreply.github.com>
Co-authored-by: Olaleye Blessing <Olayinkablexxy@gmail.com>
Co-authored-by: ktssr <31731919+ktssr@users.noreply.github.com>
Co-authored-by: Harkunwar Kochar <10580591+Harkunwar@users.noreply.github.com>
Co-authored-by: vasanthkumar <42891954+vasanth9@users.noreply.github.com>
Co-authored-by: Floran Hachez <floran.hachez@gmail.com>
Co-authored-by: Jatin <96469998+JatinSharma32@users.noreply.github.com>
* feat(stability): update snapshot
* Update components/Api/Stability/index.tsx
Co-authored-by: Jithil P Ponnan <MrJithil@users.noreply.github.com>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
* code format
* Update components/Api/Stability/index.tsx
Co-authored-by: Jithil P Ponnan <MrJithil@users.noreply.github.com>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
---------
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
Signed-off-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
Co-authored-by: Guilherme Araújo <guilherme.araujo@maxxidata.com>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: HinataKah0 <128208841+HinataKah0@users.noreply.github.com>
Co-authored-by: Olaleye Blessing <Olayinkablexxy@gmail.com>
Co-authored-by: ktssr <31731919+ktssr@users.noreply.github.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <sandeep.konala@knacksystems.com>
Co-authored-by: Harkunwar Kochar <10580591+Harkunwar@users.noreply.github.com>
Co-authored-by: vasanthkumar <42891954+vasanth9@users.noreply.github.com>
Co-authored-by: Jatin <96469998+JatinSharma32@users.noreply.github.com>
Co-authored-by: Aymen Naghmouchi <aymenadvance@gmail.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Yagiz Nizipli <yagiz@nizipli.com>
Co-authored-by: Nick Schonning <nschonni@gmail.com>
Co-authored-by: Floran Hachez <floran.hachez@gmail.com>
Co-authored-by: Jithil P Ponnan <MrJithil@users.noreply.github.com>
ovflowd added a commit that referenced this pull request May 18, 2023
* feat: create codebox component
* feat: add pointer to copy button
* feat: add shell support
* feat: use code element instead of pre
* chore: fix eslint issues
* refactor: remove shell codebox
* refactor: use react.fc type
* refactor: remove duplicate encoding
* refactor: use pre element for codebox
* style: some eol
* style: some eol
* refactor: use pre element for codebox
* chore: create inlinecode story
* refactor: create inline code story type
* chore: revert .browserslistrc
* test: remove duplicated test
* test: remove beginning space on code
* refactor: setup userEvent on tests
* refactor: remove useless async
* test: manual mock isomorphic-dompurify
* fix: some fixes
* chore: create multi lang story
* test: snapshot before and after switch lang
* refactor: inline mock
* refactor: use decorators
* chore: split inlinecode component
* feat: add selected lang state
* refactor: add inlinecode decorator
* refactor: remove duplicate classes
* chore: revert .browserslistrc
* test: fix tests
* chore: wip
* feat: use story component arg
* style: stories
* test: remove i18n from clipboard test
* chore: migrate pagination component (#5331)
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* chore: add a proper decorated component
* refactor: decorated inline code
* Chore(node feat) (#5338)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* (website redesign) Feat(shellbox): migration (#5234)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* fix(i18n): translation key (#5347)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* feat: add hover states
* chore: fix rebase leftovers
* chore: update snapshots
* chore: install @types/prismjs
* Update index.stories.tsx
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
* chore: remove duplicate snapshots
* chore: remove duplicate snapshot
---------
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: ktssr <31731919+ktssr@users.noreply.github.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
Co-authored-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
ovflowd added a commit that referenced this pull request May 24, 2023
* feat: create codebox component
* feat: add pointer to copy button
* feat: add shell support
* feat: use code element instead of pre
* chore: fix eslint issues
* refactor: remove shell codebox
* refactor: use react.fc type
* refactor: remove duplicate encoding
* refactor: use pre element for codebox
* style: some eol
* style: some eol
* refactor: use pre element for codebox
* chore: create inlinecode story
* refactor: create inline code story type
* chore: revert .browserslistrc
* test: remove duplicated test
* test: remove beginning space on code
* refactor: setup userEvent on tests
* refactor: remove useless async
* test: manual mock isomorphic-dompurify
* fix: some fixes
* chore: create multi lang story
* test: snapshot before and after switch lang
* refactor: inline mock
* refactor: use decorators
* chore: split inlinecode component
* feat: add selected lang state
* refactor: add inlinecode decorator
* refactor: remove duplicate classes
* chore: revert .browserslistrc
* test: fix tests
* chore: wip
* feat: use story component arg
* style: stories
* test: remove i18n from clipboard test
* chore: migrate pagination component (#5331)
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* chore: add a proper decorated component
* refactor: decorated inline code
* Chore(node feat) (#5338)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* (website redesign) Feat(shellbox): migration (#5234)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* fix(i18n): translation key (#5347)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* feat: add hover states
* chore: fix rebase leftovers
* chore: update snapshots
* chore: install @types/prismjs
* Update index.stories.tsx
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
* chore: remove duplicate snapshots
* chore: remove duplicate snapshot
---------
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: ktssr <31731919+ktssr@users.noreply.github.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
Co-authored-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
ovflowd added a commit that referenced this pull request May 24, 2023
* feat: create codebox component
* feat: add pointer to copy button
* feat: add shell support
* feat: use code element instead of pre
* chore: fix eslint issues
* refactor: remove shell codebox
* refactor: use react.fc type
* refactor: remove duplicate encoding
* refactor: use pre element for codebox
* style: some eol
* style: some eol
* refactor: use pre element for codebox
* chore: create inlinecode story
* refactor: create inline code story type
* chore: revert .browserslistrc
* test: remove duplicated test
* test: remove beginning space on code
* refactor: setup userEvent on tests
* refactor: remove useless async
* test: manual mock isomorphic-dompurify
* fix: some fixes
* chore: create multi lang story
* test: snapshot before and after switch lang
* refactor: inline mock
* refactor: use decorators
* chore: split inlinecode component
* feat: add selected lang state
* refactor: add inlinecode decorator
* refactor: remove duplicate classes
* chore: revert .browserslistrc
* test: fix tests
* chore: wip
* feat: use story component arg
* style: stories
* test: remove i18n from clipboard test
* chore: migrate pagination component (#5331)
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* chore: add a proper decorated component
* refactor: decorated inline code
* Chore(node feat) (#5338)
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* (website redesign) Feat(shellbox): migration (#5234)
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
* fix(i18n): translation key (#5347)
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
* feat: add hover states
* chore: fix rebase leftovers
* chore: update snapshots
* chore: install @types/prismjs
* Update index.stories.tsx
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
* chore: remove duplicate snapshots
* chore: remove duplicate snapshot
---------
Signed-off-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: ktssr <31731919+ktssr@users.noreply.github.com>
Co-authored-by: Teja Sai Sandeep Reddy Konala <tejasaisandeepreddykonala@MacBook-Pro.local>
Co-authored-by: Claudio Wunder <cwunder@gnome.org>
Co-authored-by: Wai.Tung <maledong_public@foxmail.com>
Co-authored-by: Claudio Wunder <cwunder@hubspot.com>
Co-authored-by: Augustin Mauroy <augustin.mauroy@outlook.fr>
Co-authored-by: Manish Kumar ⛄ <manishprivet@protonmail.com>
Co-authored-by: Shanmughapriyan S <priyanshan03@gmail.com>
Co-authored-by: Michael Esteban <mickel13@gmail.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@ktssr@targos@ovflowd@mikeesto@shanpriyan