Skip to content

Godot: Document disabled-in-editor-play option - #14438

Merged
limbonaut merged 3 commits into
masterfrom
godot-option-disabled-in-editor-play
Aug 11, 2025
Merged

Godot: Document disabled-in-editor-play option#14438
limbonaut merged 3 commits into
masterfrom
godot-option-disabled-in-editor-play

Conversation

@limbonaut

@limbonautlimbonaut commented Jul 23, 2025

Copy link
Copy Markdown
Collaborator

DESCRIBE YOUR PR

Add documentation for disabled_in_editor_play option added in:

Remove option according to changes:

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.

  • Urgent deadline (GA date, etc.):
  • Other deadline:
  • None: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've added an urgent due date to it.
    Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

LEGAL BOILERPLATE

Look, I get it. The entity doing business as "Sentry" was incorporated in the State of Delaware in 2015 as Functional Software, Inc. and is gonna need some rights from me in order to utilize my contributions in this here PR. So here's the deal: I retain all rights, title and interest in and to my contributions, and by keeping this boilerplate intact I confirm that Sentry can use, modify, copy, and redistribute my contributions, under Sentry's choice of terms.

EXTRA RESOURCES

@vercel

vercelBot commented Jul 23, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

NameStatusPreviewCommentsUpdated (UTC)
sentry-docs✅ Ready (Inspect)Visit Preview💬 Add feedbackJul 29, 2025 9:27am
1 Skipped Deployment
NameStatusPreviewCommentsUpdated (UTC)
develop-docs⬜️ Ignored (Inspect)Visit PreviewJul 29, 2025 9:27am

Comment threaddocs/platforms/godot/configuration/options.mdx Outdated
@codeowner-assignment
codeowner-assignmentBot requested a review from a teamJuly 25, 2025 08:11
@codecov

codecovBot commented Jul 25, 2025

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 765 bytes (0.0%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-client-array-push9.84MB-6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs12.24MB771 bytes (0.01%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
1729.js-3 bytes1.75MB-0.0%
../instrumentation.js-3 bytes1.08MB-0.0%
9523.js-3 bytes1.05MB-0.0%
../app/[[...path]]/page.js.nft.json260 bytes724.85kB0.04%
../app/platform-redirect/page.js.nft.json260 bytes724.76kB0.04%
../app/sitemap.xml/route.js.nft.json260 bytes722.24kB0.04%
view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
static/chunks/pages/_app-*.js-3 bytes880.61kB-0.0%
static/chunks/1831-*.js-3 bytes423.03kB-0.0%
server/middleware-*.js-5.55kB1.0kB-84.74%
server/middleware-*.js5.55kB6.55kB555.3% ⚠️
static/Kd-*.js(New)77 bytes77 bytes100.0% 🚀
static/Kd-*.js(New)684 bytes684 bytes100.0% 🚀
static/KsM_DkG9ra1PJ5Is7aPF4/_buildManifest.js(Deleted)-684 bytes0 bytes-100.0% 🗑️
static/KsM_DkG9ra1PJ5Is7aPF4/_ssgManifest.js(Deleted)-77 bytes0 bytes-100.0% 🗑️

@limbonaut

Copy link
Copy Markdown
CollaboratorAuthor

We removed disabled-in-editor option in favor of disabling SDK in Godot editor by default. I've updated this PR to reflect these changes.

@limbonaut
limbonaut merged commit 4ee75ff into masterAug 11, 2025
13 checks passed
@limbonaut
limbonaut deleted the godot-option-disabled-in-editor-play branch August 11, 2025 14:35
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 27, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@limbonaut@bruno-garcia@vaind@coolguyzone