Skip to content

fix(js/browser): Amend user IP + sendDefaultPii change to v10 migration guide - #14609

Merged
Lms24 merged 1 commit into
masterfrom
lms/fix-js-browser-migration-sendDefaultPii-ammend
Aug 12, 2025
Merged

fix(js/browser): Amend user IP + sendDefaultPii change to v10 migration guide#14609
Lms24 merged 1 commit into
masterfrom
lms/fix-js-browser-migration-sendDefaultPii-ammend

Conversation

@Lms24

@Lms24Lms24 commented Aug 11, 2025

Copy link
Copy Markdown
Member

Mentions a fix to be introduced in 10.4.0 in the migration guide given it has quite a lot of impact, albeit belated (should have been like this since v9).

This change is now mentioned in:

  • 10.4.0 changelog
  • in-repo migration guide
  • in-docs migration guide (this PR)

@vercel

vercelBot commented Aug 11, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

NameStatusPreviewCommentsUpdated (UTC)
sentry-docs✅ Ready (Inspect)Visit Preview💬 Add feedbackAug 11, 2025 9:30am
1 Skipped Deployment
NameStatusPreviewCommentsUpdated (UTC)
develop-docs⬜️ Ignored (Inspect)Visit PreviewAug 11, 2025 9:30am

@Lms24Lms24 changed the title fix(js/browser): Ammend user IP + sendDefaultPii change in migration guidefix(js/browser): Ammend user IP + sendDefaultPii change to v10 migration guideAug 11, 2025
@Lms24Lms24 self-assigned this Aug 11, 2025
@Lms24
Lms24 requested review from andreiborza and mydeaAugust 11, 2025 09:13
@Lms24Lms24 changed the title fix(js/browser): Ammend user IP + sendDefaultPii change to v10 migration guidefix(js/browser): Amend user IP + sendDefaultPii change to v10 migration guideAug 11, 2025
@codecov

codecovBot commented Aug 11, 2025

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 870 bytes (0.0%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-client-array-push9.85MB-6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs12.38MB876 bytes (0.01%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
1729.js-3 bytes1.77MB-0.0%
../instrumentation.js-3 bytes1.1MB-0.0%
9523.js-3 bytes1.08MB-0.0%
../app/[[...path]]/page.js.nft.json295 bytes745.38kB0.04%
../app/platform-redirect/page.js.nft.json295 bytes745.29kB0.04%
../app/sitemap.xml/route.js.nft.json295 bytes742.76kB0.04%
view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
static/chunks/pages/_app-*.js-3 bytes882.71kB-0.0%
static/chunks/8321-*.js-3 bytes425.87kB-0.0%
static/kzR_z16xnwCCtDFQAZVRX/_buildManifest.js(New)684 bytes684 bytes100.0% 🚀
static/kzR_z16xnwCCtDFQAZVRX/_ssgManifest.js(New)77 bytes77 bytes100.0% 🚀
static/OyMpO1gIKXV2rF4F-*.js(Deleted)-684 bytes0 bytes-100.0% 🗑️
static/OyMpO1gIKXV2rF4F-*.js(Deleted)-77 bytes0 bytes-100.0% 🗑️

@Lms24
Lms24 merged commit 3271c86 into masterAug 12, 2025
13 checks passed
@Lms24
Lms24 deleted the lms/fix-js-browser-migration-sendDefaultPii-ammend branch August 12, 2025 11:41
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 28, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@Lms24@andreiborza