Skip to content

Unreal Engine: Update plugin docs - #14624

Merged
tustanivsky merged 3 commits into
masterfrom
unreal/docs-update
Aug 12, 2025
Merged

Unreal Engine: Update plugin docs#14624
tustanivsky merged 3 commits into
masterfrom
unreal/docs-update

Conversation

@tustanivsky

@tustanivskytustanivsky commented Aug 12, 2025

Copy link
Copy Markdown
Collaborator

Key Changes:

@vercel

vercelBot commented Aug 12, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

ProjectDeploymentPreviewCommentsUpdated (UTC)
sentry-docsReadyPreview💬 Add feedbackAug 12, 2025 9:10am
1 Skipped Deployment
ProjectDeploymentPreviewCommentsUpdated (UTC)
develop-docs⬜️ IgnoredPreviewAug 12, 2025 9:10am

@codeowner-assignment
codeowner-assignmentBot requested a review from a teamAugust 12, 2025 08:49
@codecov

codecovBot commented Aug 12, 2025

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 570 bytes (0.0%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-client-array-push9.9MB-6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs12.43MB576 bytes (0.0%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
static/chunks/pages/_app-*.js-3 bytes882.71kB-0.0%
static/chunks/8321-*.js-3 bytes425.87kB-0.0%
static/l3QvimYhYt76DFIe4ge77/_buildManifest.js(New)684 bytes684 bytes100.0% 🚀
static/l3QvimYhYt76DFIe4ge77/_ssgManifest.js(New)77 bytes77 bytes100.0% 🚀
static/RjNtJH0w1AghyGK6SwXzs/_buildManifest.js(Deleted)-684 bytes0 bytes-100.0% 🗑️
static/RjNtJH0w1AghyGK6SwXzs/_ssgManifest.js(Deleted)-77 bytes0 bytes-100.0% 🗑️
view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
1729.js-3 bytes1.78MB-0.0%
../instrumentation.js-3 bytes1.1MB-0.0%
9523.js-3 bytes1.08MB-0.0%
../app/[[...path]]/page.js.nft.json195 bytes747.6kB0.03%
../app/platform-redirect/page.js.nft.json195 bytes747.51kB0.03%
../app/sitemap.xml/route.js.nft.json195 bytes744.99kB0.03%

@tustanivsky
tustanivsky merged commit 96dc6e8 into masterAug 12, 2025
13 checks passed
@tustanivsky
tustanivsky deleted the unreal/docs-update branch August 12, 2025 13:26
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 28, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@tustanivsky@mujacica