Skip to content

feat: Update min js sdk version for logging and update console logging snippet - #14614

Merged
AbhiPrasad merged 1 commit into
masterfrom
abhi-js-log-min-version
Aug 12, 2025
Merged

feat: Update min js sdk version for logging and update console logging snippet#14614
AbhiPrasad merged 1 commit into
masterfrom
abhi-js-log-min-version

Conversation

@AbhiPrasad

Copy link
Copy Markdown
Contributor

In 9.41.0 of the JS SDK, we dropped the experimental enableLogs option. Given we document snippets without the experimental option everywhere, let's bump the min version in the docs: https://github.com/getsentry/sentry-javascript/blob/develop/CHANGELOG.md#9410

This is the version we will lock in for GA.

While I was here, I also cleaned up the consoleLoggingIntegration so that we follow the order log < warn < error.

@AbhiPrasad
AbhiPrasad requested a review from a teamAugust 11, 2025 19:32
@AbhiPrasadAbhiPrasad self-assigned this Aug 11, 2025
@AbhiPrasad
AbhiPrasad requested review from Lms24 and RulaKhaled and removed request for a teamAugust 11, 2025 19:32
@vercel

vercelBot commented Aug 11, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

ProjectStatusPreviewCommentsUpdated (UTC)
sentry-docsReadyVisit Preview💬 Add feedbackAug 11, 2025 7:56pm
1 Skipped Deployment
ProjectStatusPreviewCommentsUpdated (UTC)
develop-docs⬜️ IgnoredAug 11, 2025 7:56pm

@codecov

codecovBot commented Aug 11, 2025

Copy link
Copy Markdown

Bundle Report

Changes will decrease total bundle size by 15 bytes (-0.0%) ⬇️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-client-array-push9.9MB-6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs12.42MB-9 bytes (-0.0%) ⬇️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
1729.js-3 bytes1.78MB-0.0%
../instrumentation.js-3 bytes1.1MB-0.0%
9523.js-3 bytes1.08MB-0.0%
view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
static/chunks/pages/_app-*.js-3 bytes882.71kB-0.0%
static/chunks/8321-*.js-3 bytes425.87kB-0.0%
server/middleware-*.js-5.55kB1.0kB-84.74%
server/middleware-*.js5.55kB6.55kB555.3% ⚠️
static/_S5dd24vz4V5SCT8ALHLD/_buildManifest.js(New)684 bytes684 bytes100.0% 🚀
static/_S5dd24vz4V5SCT8ALHLD/_ssgManifest.js(New)77 bytes77 bytes100.0% 🚀
static/zu9C8XAvX7kE6hGRuq9aK/_buildManifest.js(Deleted)-684 bytes0 bytes-100.0% 🗑️
static/zu9C8XAvX7kE6hGRuq9aK/_ssgManifest.js(Deleted)-77 bytes0 bytes-100.0% 🗑️

@AbhiPrasad
AbhiPrasad merged commit 66bf06f into masterAug 12, 2025
13 checks passed
@AbhiPrasad
AbhiPrasad deleted the abhi-js-log-min-version branch August 12, 2025 12:48
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 28, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@AbhiPrasad@Lms24