Skip to content

ref(ourlogs): Add log size limit - #14635

Merged
k-fish merged 1 commit into
masterfrom
ref/ourlogs/logs-280-add-size-limit-for-logs
Aug 12, 2025
Merged

ref(ourlogs): Add log size limit#14635
k-fish merged 1 commit into
masterfrom
ref/ourlogs/logs-280-add-size-limit-for-logs

Conversation

@k-fish

Copy link
Copy Markdown
Member

@k-fish
k-fish requested a review from AbhiPrasadAugust 12, 2025 16:00
@linear

linearBot commented Aug 12, 2025

Copy link
Copy Markdown

@vercel

vercelBot commented Aug 12, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

ProjectDeploymentPreviewCommentsUpdated (UTC)
develop-docsReadyPreviewCommentAug 12, 2025 4:06pm
1 Skipped Deployment
ProjectDeploymentPreviewCommentsUpdated (UTC)
sentry-docs⬜️ IgnoredAug 12, 2025 4:06pm

@k-fish
k-fish merged commit 6052cb8 into masterAug 12, 2025
12 checks passed
@k-fish
k-fish deleted the ref/ourlogs/logs-280-add-size-limit-for-logs branch August 12, 2025 16:41
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 28, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@k-fish@AbhiPrasad