Skip to content

feat(js): Add ignoreSpans documentation to Options page - #14572

Merged
Lms24 merged 2 commits into
masterfrom
lms/feat-js-ignorespans-option
Aug 12, 2025
Merged

feat(js): Add ignoreSpans documentation to Options page#14572
Lms24 merged 2 commits into
masterfrom
lms/feat-js-ignorespans-option

Conversation

@Lms24

@Lms24Lms24 commented Aug 6, 2025

Copy link
Copy Markdown
Member

DESCRIBE YOUR PR

Adds documentation for the new ignoreSpans option to be released with SDK version 10.2.0.

SDK PR: getsentry/sentry-javascript#16820
Develop Docs spec change: #14571

closes#14569

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.

  • Urgent deadline (GA date, etc.):
  • Other deadline:
  • None: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've added an urgent due date to it.
    Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

LEGAL BOILERPLATE

Look, I get it. The entity doing business as "Sentry" was incorporated in the State of Delaware in 2015 as Functional Software, Inc. and is gonna need some rights from me in order to utilize my contributions in this here PR. So here's the deal: I retain all rights, title and interest in and to my contributions, and by keeping this boilerplate intact I confirm that Sentry can use, modify, copy, and redistribute my contributions, under Sentry's choice of terms.

EXTRA RESOURCES

@vercel

vercelBot commented Aug 6, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

ProjectDeploymentPreviewCommentsUpdated (UTC)
sentry-docsReadyPreviewCommentAug 12, 2025 0:04am
1 Skipped Deployment
ProjectDeploymentPreviewCommentsUpdated (UTC)
develop-docs⬜️ IgnoredPreviewAug 12, 2025 0:04am

@Lms24
Lms24 requested review from a team, chargome and mydea and removed request for a teamAugust 6, 2025 15:27
@codecov

codecovBot commented Aug 6, 2025

Copy link
Copy Markdown

Bundle Report

Changes will increase total bundle size by 375 bytes (0.0%) ⬆️. This is within the configured threshold ✅

Detailed changes
Bundle nameSizeChange
sentry-docs-client-array-push9.85MB-6 bytes (-0.0%) ⬇️
sentry-docs-server-cjs12.37MB381 bytes (0.0%) ⬆️

Affected Assets, Files, and Routes:

view changes for bundle: sentry-docs-server-cjs

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
1729.js-3 bytes1.77MB-0.0%
../instrumentation.js-3 bytes1.1MB-0.0%
9523.js-3 bytes1.08MB-0.0%
../app/[[...path]]/page.js.nft.json130 bytes741.53kB0.02%
../app/platform-redirect/page.js.nft.json130 bytes741.45kB0.02%
../app/sitemap.xml/route.js.nft.json130 bytes738.92kB0.02%
view changes for bundle: sentry-docs-client-array-push

Assets Changed:

Asset NameSize ChangeTotal SizeChange (%)
static/chunks/pages/_app-*.js-3 bytes882.71kB-0.0%
static/chunks/8321-*.js-3 bytes425.87kB-0.0%
server/middleware-*.js-5.55kB1.0kB-84.74%
server/middleware-*.js5.55kB6.55kB555.3% ⚠️
static/p8-*.js(New)684 bytes684 bytes100.0% 🚀
static/p8-*.js(New)77 bytes77 bytes100.0% 🚀
static/tJbTk21BgPWBRWG9KpViv/_buildManifest.js(Deleted)-684 bytes0 bytes-100.0% 🗑️
static/tJbTk21BgPWBRWG9KpViv/_ssgManifest.js(Deleted)-77 bytes0 bytes-100.0% 🗑️

Comment threaddocs/platforms/javascript/common/configuration/options.mdx Outdated

@coolguyzonecoolguyzone left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

Co-authored-by: Alex Krawiec <alex.krawiec@sentry.io>
@Lms24
Lms24 enabled auto-merge (squash) August 12, 2025 11:42
@Lms24
Lms24 merged commit 7b1731c into masterAug 12, 2025
13 checks passed
@Lms24
Lms24 deleted the lms/feat-js-ignorespans-option branch August 12, 2025 12:04
codyde added a commit that referenced this pull request Aug 13, 2025
* 'master' of https://github.com/getsentry/sentry-docs: (22 commits)
Fix incorrectly placed comma (#14592)
fix(devservices): Typo in devservices docs (#14636)
ref(ourlogs): Add log size limit (#14635)
Document undocumented php prefix option (#14634)
getsentry/relay@7958273
Unreal Engine: Update plugin docs (#14624)
feat: Update min js sdk version for logging and update console logging snippet (#14614)
feat: [Unreal] Add transaction and span set_data documentation (#14599)
feat(js): Add `ignoreSpans` documentation to Options page (#14572)
fix(js/browser): Amend user IP + `sendDefaultPii` change to v10 migration guide (#14609)
fix: Fix broken AI banner due to invalid syntax (#14627)
feat(partners)add-enablement-docs (#14095)
Fix image path issue & add test coverage for new lightbox (#14616)
feat(seer): update GHE setup to work with seer (#14617)
fix(replay): Document searchable properties `screen` and alias `screens` (#14603)
Update image handling (#14564)
Update using-hooks.mdx (#14605)
Godot: Document `disabled-in-editor-play` option (#14438)
fix(Docs) redirect and deprecate old metrics pages (#14602)
fix(relay): Update projects.mdx (#14606)
...
@github-actionsgithub-actionsBot locked and limited conversation to collaborators Aug 28, 2025
Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Add ignoreSpans documentation to JS SDK docs

3 participants

@Lms24@AbhiPrasad@coolguyzone